)]}'
{"id":"openstack%2Fwatcher~960165","triplet_id":"openstack%2Fwatcher~master~I68a7d35295b83067c626f85637141591b3d07f22","project":"openstack/watcher","branch":"master","attention_set":{},"removed_from_attention_set":{"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2026-09-10 09:26:07.000000000","reason":"\u003cGERRIT_ACCOUNT_34452\u003e replied on the change","reason_account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-08-01 00:45:14.000000000","reason":"removed on reply"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 09:33:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I68a7d35295b83067c626f85637141591b3d07f22","subject":"Drop PasteDeploy","status":"MERGED","created":"2025-09-08 17:50:17.000000000","updated":"2026-09-10 09:34:24.000000000","submitted":"2026-09-10 09:33:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960165","meta_rev_id":"f40e1f0396b77d7e3f67f0d8c772841021f7975d","_number":960165,"virtual_id_number":960165,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 09:33:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2026-08-24 12:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2026-09-10 09:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":-1,"date":"2026-08-24 06:15:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2026-09-10 09:26:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}],"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-08 18:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-08 20:57:26.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-10-27 16:33:43.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"REVIEWER"},{"updated":"2026-08-01 00:34:45.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-08-24 12:02:55.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"81d027d37781fc56786af59335a964276862d030","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-08 17:50:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb80bcccf1b0cd66e920a94ef680874d8f18c020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 18:51:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26ad61efb5d242e0bce5fc256cb5520e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/362d97565fdb44d382ce49e93ae8934e : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41a0d1939f2940269d07beedbcb333e4 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18eef7e4eace44dd81f54fdf1ea4efcf : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9eeff98e95c4457da3b82330e00a2016 : SUCCESS in 5m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0230c5df6ffe46cd93f5f7403e4e395d : SUCCESS in 5m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45af2a4d17984255aa1b5624f197badd : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b022dadfcf64d50a5d20230bc7028cc : SUCCESS in 9m 20s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/07bacb0b4730442484783b6724487239 : SUCCESS in 5m 27s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/9e97fe2a6b0c4da6ae0d8389f4cdf8fb : SUCCESS in 33m 18s\n- watcher-tempest-strategies https://zuul.opendev.org/t/openstack/build/4199bc978ff74e9e9244e616f86f66bc : SUCCESS in 39m 24s\n- watcher-tempest-actuator https://zuul.opendev.org/t/openstack/build/710c740841bc45fea53afdba96f031d5 : SUCCESS in 47m 26s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/74307f32ae85430c86d6953690ad4f0e : SUCCESS in 36m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1d17421e9d0d009a19c390a7572471920a9a766f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-08 20:57:26.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/6b5b8d8bf2324b339b690593ce7c9583\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/3cadba8ce1f6426da130d93f7cf704f2 : SUCCESS in 3h 05m 51s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4aa877e1e52345c3be461cb6c35bfc83 : SUCCESS in 2h 08m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54d4cb4404eadc6a308b583739532b4cb8489611","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-10-27 16:33:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"50d91961841501143615b02a5a7b7df07f79670c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-01 00:34:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1bbfc0308eae397ca4f7951e12a4a3a6cd07c846","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-01 00:34:45.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3d2c5bbb2388de73c19be4c2c2b736f99c4fbf7b","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-01 00:44:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/dcd8a0b06b28492880aab8e4b1cfe42e\n\n- teim-code-review https://zuul.teim.app/t/main/build/d93dab1f79c74fb5bf7933a6544f7433 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2},{"id":"28e6644c1bbe4347507b998425f1aea861352b27","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-01 00:45:14.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/23287d64f1b447a9a093dc71bc37eb09\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/759d857b1d0647cd9e0a03ff7ffa04f6 : FAILURE in 8m 57s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/4e2152c711c840a4b4c7a8c4fc979adc : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":2},{"id":"893c0ee80bc2baace1d2d24013e20e03b91e31c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-01 00:49:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b328d623a0b54852bc4449783dcf150e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/77e247f8cb5b42e2bb4c629e8f2ebdbb : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2382823e53d4ff5a0f1f374da30cab5 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2220b04b41484640ac4c5ec5989ba0c2 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7774ac4e4fe6441680efffba3f8099f9 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cfcd112be54d42a587d5056ec0699119 : SUCCESS in 5m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/94e489e93a554e92962860ecb9ae69b3 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aea584ab463841c49b5354ad7b282cb2 : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8b8b9531d8a4ffb9df9c81a072be513 : SUCCESS in 4m 02s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/bcb563e16fea4f0f9c117132053f2e8c : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5374bb697f562c36011cd610286823667df8e85b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-24 06:04:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3c209562acd3eafd887b5c4a5bf37b9799b28ad3","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-24 06:04:54.000000000","message":"Patch Set 3:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e338d411e1afc8f4b2289054c09aec49a64f6592","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-24 06:15:41.000000000","message":"Patch Set 3: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ab03342777d040d9952ab2e3dba8aee6\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/0efd989c855c49aa86376517bddb09b3 : FAILURE in 8m 10s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/63b7f9cb879e445aab4359a393e5b821 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":3},{"id":"9e097808f3f03d9375dd3ed840c80b5c1f334614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 06:15:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/807907c5ad414b92a5bf7a204fd2a1c7\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b5b0dd3bdf3f445ea5ff13334bb93034 : SUCCESS in 1m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e860c5d73891415289a5f1fad1b3c942 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/320476889fa348d3a771d8f8d3be77aa : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8242b9fd65224385a8e298a6647cb661 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1766d9bc70764e33bd7b56d1fa2564d8 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/098206899fe64860a0d32883886f8414 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6626529b19440a38f18281d0495020a : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f7f1f6ddfd74deab7f5623ed4b0399e : SUCCESS in 2m 53s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/ab7a49b7cd2b41faa4c9a631ce7a3d38 : SUCCESS in 4m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/78be6de9cab940e8bc83bdb48450ecef : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c8ba73249a4c58bca03efe87586978e7631d684f","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-24 07:30:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/bb3795eae0b140119139671c7647f849\n\n- teim-code-review https://zuul.teim.app/t/main/build/e3778024dff84faa968b99a4707a151f : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":3},{"id":"ba40c0f9437793e851fd55dd96d862c51a9629a4","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-08-24 12:02:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a54fc0adeb22740e064d04f5d91935ffaebf9e32","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-09-10 09:26:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3b7ada1fa9319acbc882eb37ba64150c81042731","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 09:26:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"250003dc40204e5b1089a9fd1c714227251abfe9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 09:33:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e0d92de56a4c7b82e0a5ecb41e89af\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c91c7d2eaa504a069e67879e16a9daa3 : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44c7785c6803410ebecb0d0b11666792 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f96a98cf2d8947b1a38bee6789bae84a : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a777487d6774abc88912d421b4aa2cf : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/676a7898889142d782e57f04208da123 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1151f09945b43128e36ca02bb428c8a : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cd45b4dd16b2488ea9ef237573f6959c : SUCCESS in 5m 38s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/107c4c8348094aeab08287139bf3a446 : SUCCESS in 3m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7cd11f4facc84b53bffcf949d04bc7b8 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":3},{"id":"327686b6dfb922b07a7f1dac758960eeca618db5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 09:33:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f40e1f0396b77d7e3f67f0d8c772841021f7975d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 09:34:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/205c608c2d8346b180354b0d9403eabe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5618af01d76b4d6f83282e29aec3bfa7 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fea4ca476fa944243790089a2da4f837cdb92a77","revisions":{"2eef50cd4fbfb714c5a977775f75449232dcae8b":{"kind":"REWORK","_number":1,"created":"2025-09-08 17:50:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/960165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/65/960165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/65/960165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/65/960165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/65/960165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/65/960165/1"}}},"commit":{"parents":[{"commit":"b1aad46209fd4302fc413146a15c5c2406d3dd60","subject":"Merge \"Check result of retype action based on type and status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/b1aad46209fd4302fc413146a15c5c2406d3dd60"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-08 17:48:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-08 17:50:05.000000000","tz":540},"subject":"Drop PasteDeploy","message":"Drop PasteDeploy\n\nIt\u0027s not used directly by watcher, because watcher uses pecan instead.\nNote that the library is still pulled as a requirement of oslo.service.\n\nChange-Id: I68a7d35295b83067c626f85637141591b3d07f22\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/2eef50cd4fbfb714c5a977775f75449232dcae8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/2eef50cd4fbfb714c5a977775f75449232dcae8b"}]},"branch":"refs/heads/master"},"c81483331d978b7197148891405e4d43467898a9":{"kind":"REWORK","_number":2,"created":"2026-08-01 00:34:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/960165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/65/960165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/65/960165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/65/960165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/65/960165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/65/960165/2"}}},"commit":{"parents":[{"commit":"4bca4144c8fadd2fbcb4720253488b35e94bbeba","subject":"Merge \"Replace python-cinderclient with openstacksdk\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/4bca4144c8fadd2fbcb4720253488b35e94bbeba"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-08 17:48:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-01 00:34:29.000000000","tz":540},"subject":"Drop PasteDeploy","message":"Drop PasteDeploy\n\nIt\u0027s not used directly by watcher, because watcher uses pecan instead.\nNote that the library is still pulled as a requirement of oslo.service.\n\nChange-Id: I68a7d35295b83067c626f85637141591b3d07f22\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/c81483331d978b7197148891405e4d43467898a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/c81483331d978b7197148891405e4d43467898a9"}]},"branch":"refs/heads/master"},"fea4ca476fa944243790089a2da4f837cdb92a77":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-24 06:04:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/960165/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/65/960165/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/65/960165/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/65/960165/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/65/960165/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/65/960165/3"}}},"commit":{"parents":[{"commit":"ec06744bffd4ba858da7f4996cc872723dfbde70","subject":"Merge \"Add Nova/Placement emulators and host_maintenance functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/ec06744bffd4ba858da7f4996cc872723dfbde70"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-08 17:48:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 06:04:45.000000000","tz":0},"subject":"Drop PasteDeploy","message":"Drop PasteDeploy\n\nIt\u0027s not used directly by watcher, because watcher uses pecan instead.\nNote that the library is still pulled as a requirement of oslo.service.\n\nChange-Id: I68a7d35295b83067c626f85637141591b3d07f22\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/fea4ca476fa944243790089a2da4f837cdb92a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/fea4ca476fa944243790089a2da4f837cdb92a77"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"4bca4144c8fadd2fbcb4720253488b35e94bbeba","ours":"c81483331d978b7197148891405e4d43467898a9","theirs":"ec06744bffd4ba858da7f4996cc872723dfbde70","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
