)]}'
{"id":"openstack%2Fwatcher~966226","triplet_id":"openstack%2Fwatcher~master~I85926cdd1e4393be42fe6d543af50bf2948b5ce6","project":"openstack/watcher","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2025-12-30 17:05:28.000000000","reason":"Change was submitted"},"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2025-12-30 17:05:28.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-12-30 15:44:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"28006":{"account":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"last_update":"2025-12-01 01:49:49.000000000","reason":"removed on reply"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-12-29 15:49:57.000000000","reason":"removed on reply"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2025-12-30 17:05:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I85926cdd1e4393be42fe6d543af50bf2948b5ce6","subject":"Adds support for threading mode in applier","status":"MERGED","created":"2025-11-05 20:46:32.000000000","updated":"2025-12-30 17:07:03.000000000","submitted":"2025-12-30 17:05:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":40,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966226-eventlet-removal","meta_rev_id":"468104b085b9abef099684dfdc0c475233a15d3a","_number":966226,"virtual_id_number":966226,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-30 17:05:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2025-12-30 15:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":1,"date":"2025-12-29 15:49:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2025-12-30 15:44:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}],"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-05 22:04:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-05 23:27:31.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-11-12 13:48:09.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2025-11-17 19:56:26.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-11-17 19:56:26.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"REVIEWER"},{"updated":"2025-11-19 10:26:31.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"2b6bb98688ab2a51479d6bb1ca8e45b781185d37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-11-05 20:46:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f2e586c601853992411ba8e29ca87ceed6941e8","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-11-05 20:46:47.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25b1133e5fda5156aa8835c786f16376958d30cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 22:04:48.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16fa67c80ca9494f8dc83ce21e6f0992\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/917c147e827e4cfa87f5c5aabdb6d4ee : SUCCESS in 5m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c24a1cca42fa49aaa4973d2ebb7fca4e : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba1dae72c3624d1bb519f87844fc9477 : FAILURE in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/051c23c5088b45f2acfd50a65806ea30 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf846a7085204124b6018dbd977748a7 : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c887d7712084768985309e977abc332 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76b1b838ce0a4415a9bceccda32b1e64 : SUCCESS in 4m 06s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/f827ad85b1294c43bae471abcbb69d2d : FAILURE in 3m 51s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/3d6c9667df8c45b0854b26f306020cdd : SUCCESS in 18m 52s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/cd97b0049f34491eb33a215bf3b7da8d : SUCCESS in 46m 20s\n- watcher-tempest-strategies https://zuul.opendev.org/t/openstack/build/edbe29800bb64748bb68391e51eff195 : SUCCESS in 38m 36s\n- watcher-tempest-actuator https://zuul.opendev.org/t/openstack/build/0b9095e12d39420499d5dae6ac4d450d : SUCCESS in 44m 37s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/935fe70756fe47759d8e640d08ea3a56 : SUCCESS in 29m 11s\n- watcher-prometheus-integration https://zuul.opendev.org/t/openstack/build/cee9b69fc65f4841aefb3b83ba8e286d : SUCCESS in 56m 14s\n- watcher-prometheus-integration-threading https://zuul.opendev.org/t/openstack/build/a4b2c33f956044a6b6f79d2b8ab36ea2 : SUCCESS in 1h 12m 51s\n- watcher-aetos-integration https://zuul.opendev.org/t/openstack/build/e3ad85a7c78a49c094a8ca48344391cd : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2fd9a807ca31083c8a95ef6b06dd31953ef7f69a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-05 23:27:31.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/4697b91e839d4969b861cc6d5785864a\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b350a7e83ad844b891b4b001e021dcd9 : SUCCESS in 2h 39m 41s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/0dd441c671f44899812a2d6865698d5d : SUCCESS in 1h 53m 18s","accounts_in_message":[],"_revision_number":1},{"id":"99f382dd474a0de932296c258855a5b61726a41b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-11-12 13:47:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e7ec361f783d264899c01bb0ddfbfb2d64cf3f6b","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-11-12 13:48:09.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36f4c95aabfcfef7af21bfd93b2d407901fa801e","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-11-12 14:01:13.000000000","message":"Patch Set 2:\n\n(9 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/d33a93f786fe4574a1657161610ec2bc\n\n- teim-code-review https://zuul.teim.app/t/main/build/6f8ebff75a3d4c958dee02dbddb6e9e3 : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":2},{"id":"5ff233404d9df6bc2be57a85a451b1f274a54409","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-11-12 16:23:11.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dcb8dc239071e3fff13c2a526d02a989283806a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 16:39:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a1e7c938e0e45a5b7f511a1fa070f6d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/adfc7f595c4342d98cc3e59e58a04c99 : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ded7ea168c854974b40b29a00e0905e3 : POST_FAILURE in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a9d0ea96d304b0fa5548f2db19f0743 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/108d862381ce46ce8a76cb88f1422baa : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3359f0cd22a4c4f8712eeabb170070a : SUCCESS in 9m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f28d314b004e3caffa7e120f472360 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86fd2cee186844c1a9deabcda6882380 : SUCCESS in 4m 41s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/366578bbccda4ff1adfbb465a555d91e : SUCCESS in 4m 09s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/7277e1404a644d259bebafbbbaaea970 : SUCCESS in 29m 07s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/e1af71cd63d945b49d757f63ddadf6c9 : SUCCESS in 46m 54s\n- watcher-tempest-strategies https://zuul.opendev.org/t/openstack/build/9f89e800c3f94912b82ac640d38c2232 : SUCCESS in 1h 07m 09s\n- watcher-tempest-actuator https://zuul.opendev.org/t/openstack/build/7a6e8bf94ebf4a9db6f9356b1a819c87 : SUCCESS in 37m 21s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/304668d7f5ba442989359e87227ba5df : SUCCESS in 30m 45s\n- watcher-prometheus-integration https://zuul.opendev.org/t/openstack/build/bfca5883c0774e328413c7e5871a3959 : SUCCESS in 1h 16m 10s\n- watcher-prometheus-integration-threading https://zuul.opendev.org/t/openstack/build/8bf41468f4744263b43f0b1d0b96cd4e : SUCCESS in 1h 14m 07s\n- watcher-aetos-integration https://zuul.opendev.org/t/openstack/build/d0b4110dc9ce4b02b713cb2f6d2eda8b : SUCCESS in 1h 17m 17s","accounts_in_message":[],"_revision_number":2},{"id":"18cc64b4aa71667c38d88c0e7606abd736498690","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-12 16:40:37.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/e4aa1052c3af4220bafa7060833517e2\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/084f6682438a4f7a9006d3c53fa9c167 : SUCCESS in 2h 51m 26s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b4c0da89cec74eeaa1dfb54f5c610212 : SUCCESS in 1h 58m 01s","accounts_in_message":[],"_revision_number":2},{"id":"ced59bced870ebff2172f20d0efa8c0d9fbc9c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-11-12 18:04:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0a4a91cd8324d3e187b42948cd9ee257d41534cf","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-11-12 18:04:18.000000000","message":"Patch Set 3:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":3},{"id":"923d5bea170ec76e5d5fa0fc0789b084bd6226dc","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-11-12 18:17:27.000000000","message":"Patch Set 3:\n\n(6 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/2e2ea8d9b6c14f00874c801988b02603\n\n- teim-code-review https://zuul.teim.app/t/main/build/9a8c8d0e11c2428e9313e2da250fcd62 : SUCCESS in 12m 58s\n\nWarning:\n  Comments left for invalid file watcher/common/executor.py","accounts_in_message":[],"_revision_number":3},{"id":"83214073b2c1019078ccd1a68681b1ffaeba753b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 19:27:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e015188e06941009f58b91e0d1c6943\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b288cd662de045f09011439fbe4d7874 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28dd3459fd1a4a138ad98091648a9f02 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e13b53bd6c3d4a2295679730d27e640d : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83ee835151704108b060490cea4545a2 : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/120d06feb5bc406da632fc8ec290f3a2 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/039e51101a4b4fad82c67832fa201777 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de28458da52f460db2336c9508621bed : SUCCESS in 4m 02s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/1f230e872cbf4c3fb0f8480a5bf8d5a2 : SUCCESS in 5m 17s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/3f511faa08e341a3b4ecb6a805ef0fb0 : SUCCESS in 30m 15s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/5e3ed913de524875a0d4d190346a2c07 : SUCCESS in 45m 59s\n- watcher-tempest-strategies https://zuul.opendev.org/t/openstack/build/c9e758157f87408ba3cc2275048a3d73 : SUCCESS in 55m 01s\n- watcher-tempest-actuator https://zuul.opendev.org/t/openstack/build/d64d6efb6d4249b282b0f6da6e813089 : SUCCESS in 37m 12s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/7ac1ab423447464f851c64f0979d0185 : SUCCESS in 24m 16s\n- watcher-prometheus-integration https://zuul.opendev.org/t/openstack/build/68f7b1370ec54c47b8eff719ea0f974c : SUCCESS in 1h 20m 31s\n- watcher-prometheus-integration-threading https://zuul.opendev.org/t/openstack/build/49495bd5c3304e5eb17d25fe43604670 : SUCCESS in 46m 25s\n- watcher-aetos-integration https://zuul.opendev.org/t/openstack/build/8579317cbd14432ea767ff5a2417cca0 : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":3},{"id":"094a1862f82d54148520ccb61cb47d963d87351a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-12 21:10:23.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/8b7e2bc6c73f4b77aa468b600cd5dc42\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/bf77f798d15a4593ab7abed131c1271d : SUCCESS in 3h 02m 01s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/6c5bb81533e742b392ccd6360d145239 : SUCCESS in 1h 57m 09s","accounts_in_message":[],"_revision_number":3},{"id":"b984dceb48a19745ee005de73fdfb5d04cf4c716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-11-13 17:29:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a3938547aba62192b51b484e8acb9a40100021c4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-11-13 17:29:16.000000000","message":"Patch Set 4:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cedf0e1149ff85ebbb592008fde24cda8176000c","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-11-13 17:45:03.000000000","message":"Patch Set 4:\n\n(7 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/5de8a40f24da42da8e13675aea58cdc4\n\n- teim-code-review https://zuul.teim.app/t/main/build/967699603890482aaa17aff159c9e0be : SUCCESS in 15m 34s\n\nWarning:\n  Comments left for invalid file watcher/common/executor.py\n  Comments left for invalid file watcher/eventlet.py","accounts_in_message":[],"_revision_number":4},{"id":"bfabe4e3daf96f35cfafa92e23d6157bcf97aebb","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-13 18:10:51.000000000","message":"Patch Set 4: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/116e3b4f09f64e29b5749e92387a3f4b\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/6a876dec2c4d4e839f1c94951e41e631 : FAILURE in 40m 22s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/837ef705fd444f60a683a7b7d53c00c1 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":4},{"id":"84a20635258bb9b696a84145700784e0553957c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 18:51:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70b61bf63a3c4591bc8a58bd004d00e2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a3ae88a1c31549739c34967382fb1580 : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f9a1f00497349e1860366c942e54485 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6df74f4e1ca24c59856af39421524c71 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/493467d35b054854bd52a4c6c18050c9 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fcaccb0295243f99868c868f456ae02 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e6fcc729824828baaf7cb367283dfa : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d94ab9cf7ef948f9899981c4c5432156 : SUCCESS in 3m 51s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/2165a83bfad74dd79aecf53291329721 : FAILURE in 5m 23s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/8e7e25f7c786448abf1cd3588fca94b9 : SUCCESS in 12m 55s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/0b74279ca52b4475aa7bc34d00c8f476 : SUCCESS in 23m 13s\n- watcher-tempest-strategies https://zuul.opendev.org/t/openstack/build/b719d359a4f8472497b05ad052f827de : SUCCESS in 1h 16m 20s\n- watcher-tempest-actuator https://zuul.opendev.org/t/openstack/build/fac4e643315443319d62f1c0169c4911 : SUCCESS in 40m 35s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/4bd73bac334f4c4f9793703d94b29835 : SUCCESS in 14m 02s\n- watcher-prometheus-integration https://zuul.opendev.org/t/openstack/build/ac7b6aead3e24cf6833b85bab44aa459 : SUCCESS in 1h 16m 29s\n- watcher-prometheus-integration-threading https://zuul.opendev.org/t/openstack/build/65a766178bef40cb80f3f73ef165f3d3 : SUCCESS in 1h 03m 09s\n- watcher-aetos-integration https://zuul.opendev.org/t/openstack/build/9ab6a260af8c477c8faf67e88f537ab6 : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":4},{"id":"8ec01cd1e077f8048d0ab32f84d5c20bfd713260","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-17 15:21:08.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d8a3d7a567e16355d9181907c1975f4a909aef87","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-11-17 15:21:26.000000000","message":"Patch Set 5:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":5},{"id":"50da133bde03ea54195d7f0eef8bdab72d41a2f2","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-11-17 15:40:50.000000000","message":"Patch Set 5:\n\n(6 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/b3c78a6b34054a17bb78f337d7961812\n\n- teim-code-review https://zuul.teim.app/t/main/build/1921d06ead6d4103a28481ab96c381e2 : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":5},{"id":"265aebf4366b8fec4ff463a2b4c4e12ffc7e1a1e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-17 15:56:35.000000000","message":"Patch Set 5: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/02eded8c73a84dbbade0ff40d30f6c10\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/45fea6629ffc4c70b18902b9c0241ea4 : FAILURE in 34m 17s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ac7e922353fe4925bd9c97034f2085cd : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":5},{"id":"aa4b09ff6118c092a207b0eac3b0383ca1aaa3ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-17 16:47:34.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"f00c3ae941bbf0eee03e399aa12ebc6e98d4b0ad","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-11-17 16:47:46.000000000","message":"Patch Set 6:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0af72529dc6ac865c1945cca28711dbaaa0126ba","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-11-17 17:04:00.000000000","message":"Patch Set 6:\n\n(5 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/54de09d14b974d7fa4aa8bf75a6a77cb\n\n- teim-code-review https://zuul.teim.app/t/main/build/4dc3ed1f08a94ffabcbf6d5ebe749667 : SUCCESS in 15m 54s","accounts_in_message":[],"_revision_number":6},{"id":"d7ed1f718c7cbdc8997afab2d78fe8e8e791dfd7","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-17 17:25:46.000000000","message":"Patch Set 6: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/7738d92a56bc4805b62680a2666d22c5\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/6bb33201b0da49aaaddd91e02fd6699c : FAILURE in 36m 55s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/01fefe4dc3124561ac689ff8313d3d82 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":6},{"id":"2f6b8c25ec5c489084ce771170033503151bc688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 18:22:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46e2c631277e4de6a44df1de156a32f5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/955c8f8af90c41a99540c3464158a679 : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43c1e2af2f9246df99fb49da81a4b543 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b2a516c2f174c2781684b3571083322 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a4c9e4b82484965bacffd06e4010a52 : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e704774eae1449beaa5167b08ca67b37 : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b68628f8bed4ca2830848b4789c2b1e : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af1a301a8ba947fda9be10038b9aeca1 : SUCCESS in 3m 24s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/f636fd4964fa4541a381327555eadad5 : SUCCESS in 3m 15s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/dbcfd6bfbc12401e9402f8dd107f1fd8 : SUCCESS in 30m 27s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/15fcc87384a444a38074bdd837d1b9e5 : SUCCESS in 45m 08s\n- watcher-tempest-strategies https://zuul.opendev.org/t/openstack/build/b45e40ceb2544d48abb2eb6a5ad60651 : SUCCESS in 55m 15s\n- watcher-tempest-actuator https://zuul.opendev.org/t/openstack/build/e27980a72d2949789ba557da43941418 : SUCCESS in 29m 20s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/752b32fa66a44b1c98aa7e1dd02109dc : SUCCESS in 30m 16s\n- watcher-prometheus-integration https://zuul.opendev.org/t/openstack/build/80b6e175c33f4898949c85f2eacc3303 : SUCCESS in 1h 14m 08s\n- watcher-prometheus-integration-threading https://zuul.opendev.org/t/openstack/build/d3ce25733d00452583e00bed9e6cc1e9 : SUCCESS in 40m 02s\n- watcher-aetos-integration https://zuul.opendev.org/t/openstack/build/7efd8f13cf4a4bc9beee38f90bf67b59 : SUCCESS in 44m 06s","accounts_in_message":[],"_revision_number":6},{"id":"b08d89c18075e437e5240ff25d5fa2c36d35aecd","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-11-18 11:26:58.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c234585c0f7b77e5a63b22d2d2bfe24956692a8f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-11-18 12:34:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e1e2a609f98b40cad3088d711624f60daf727c34","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-18 16:04:05.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ed635d47185dae1a608eee7d393f784d2f945984","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-11-18 18:04:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b5df37dd00edfe2ed045a1b25d805ee425bf2888","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-11-19 10:26:31.000000000","message":"Patch Set 6: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"96f2cc3b8dc47ff18475e364e289740a2d87c33a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-19 11:31:15.000000000","message":"Patch Set 6: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":6},{"id":"07c480e536a899237ba416f503b3c4fc2b5818aa","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-11-19 13:34:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6f0b34a819eef66e2e7de1ce0434077d387457c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-11-19 20:29:25.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"d2d4e994d8ee86cb5ec12f019e45c96af093f5f2","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-11-19 20:32:48.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e18a47af8e51a3068830b6507ba4bffc62b58cd1","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-19 21:13:08.000000000","message":"Patch Set 7: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/76583ce4e98a47b09c66d1ae2d79b4c6\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c3cd9e47e45f4e1a8daacf95b91fd659 : FAILURE in 42m 13s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/9a36bb215d364c44ae4fb729c8275975 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":7},{"id":"99696fc491d06ccd7b7d6a970bc1304e021e708c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 21:53:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba5438f415c445c684d07a83eb2f9c4e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/97afa74a392147558b76cf99b7c59249 : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81dbd0057428445eac3c7a935a65858f : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a65ccd8597b4913b3ca0fb109b05c5c : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc723cd4af674805a0ac950fbcbeb191 : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30b2cad515a246989ad40fd57d7b7a8e : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba832923c5a4825ab83b0577c9e0d00 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0d09970425e455faedfa98fc3c6c5d1 : SUCCESS in 4m 14s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/0242548d8ad24c92a4b20a2ac415760c : SUCCESS in 5m 37s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/951c2688b4a64bee85325d3bca6d6cb1 : SUCCESS in 26m 28s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/ae56d802f3ae4fde95496c8b5ef681e3 : SUCCESS in 50m 13s\n- watcher-tempest-strategies https://zuul.opendev.org/t/openstack/build/a9434c47c72b4a92be143e264e439bff : SUCCESS in 1h 09m 56s\n- watcher-tempest-actuator https://zuul.opendev.org/t/openstack/build/ac0861a69bd942c49295d7f7bc2e284d : SUCCESS in 23m 42s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/7e79644a76bf414494c162bcf37f7cb4 : SUCCESS in 30m 24s\n- watcher-prometheus-integration https://zuul.opendev.org/t/openstack/build/dc2b4685653c4690a010eb6165b49c01 : SUCCESS in 1h 12m 19s\n- watcher-prometheus-integration-threading https://zuul.opendev.org/t/openstack/build/21426e87057442529841a10c0d42f187 : SUCCESS in 1h 14m 00s\n- watcher-aetos-integration https://zuul.opendev.org/t/openstack/build/fcb4d1359a274ab996e7a02c8143265f : SUCCESS in 1h 21m 40s","accounts_in_message":[],"_revision_number":7},{"id":"843e6c0a44fd446795d01056b8d0d0eee641dc5a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-11-20 11:33:15.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"c7c767aea70c831178bf05f23b914df038fa86e8","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-11-25 11:59:51.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5a6d52ee11b31952e50fba552ddfe788a76fd2e9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-28 17:41:09.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"21d1c8416d30bad8df86554a04d84c3869b501cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 18:20:28.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"4155fc8994be890cd423a103d0f9e40fcc8dc76e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 18:51:58.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55105c76e19140de8e8e7662cd331fc3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eb4cf1e7f6924da493084bbb94a941a8 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/281e7fa6aa0b4bfe9534e0a2320f8615 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1230d27b4113411aa4107acfb7b5f9d5 : SUCCESS in 6m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20d6665464c84cb181fd1281405075ab : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25df560b10149f3a8907707a6b61464 : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbae2e63f9e24e6499f50e55738da444 : SUCCESS in 4m 34s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/45237ce5fe8e479ba11d38e7c7c63e31 : SUCCESS in 27m 26s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/662fa3717431489284d99bfc4143d17c : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":7},{"id":"706294eda5e5b67fcda8d63dfedcae59fbcb65eb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-12-01 01:49:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fadaca1a932d2e4a6da023a6e6929d2a166d86ed","tag":"autogenerated:zuul:manual-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-12-01 01:49:49.000000000","message":"Patch Set 7:\n\nStarting manual-ci jobs.","accounts_in_message":[],"_revision_number":7},{"id":"aac4520014e5c3177deff31ff487d0b69408cfa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 01:50:19.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"566442b41bad39cf98753b07f4c08d37e4073736","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-12-01 01:50:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ac43f2aeb04a2bc7df3e4e276088bd582df301ce","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-12-01 01:50:53.000000000","message":"Patch Set 7:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":7},{"id":"39b6c35afd249a0b797489dd66237ed6f6970b7b","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-12-01 02:12:37.000000000","message":"Patch Set 7:\n\nBuild failed (automatic-ci pipeline). To rerun just this ci comment\n\"teim-ci: auto\". To rerun other ci jobs comment \"recheck\".\nNote this ci does run on \"recheck\".\n\nhttps://zuul.teim.app/t/main/buildset/23cf4d2bc23043658fd840322acd3c37\n\n- teim-code-review https://zuul.teim.app/t/main/build/5cf2e15fe1154f439ea0227a4e6a3820 : FAILURE in 21m 32s","accounts_in_message":[],"_revision_number":7},{"id":"83f22003af71f0e93d80c2d71b8a51ab06b7dd59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 02:12:44.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abaceb0830564b28b7b5464233480696\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0f9e1854357a454e9e21d2a68d82b161 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3bec5dda22941319ade0d4ba3e7e995 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a536a5cd5214898a1669aaa846b91d2 : SUCCESS in 7m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34b97245836649f48f6fed74f0b5132f : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/742a946267c646ceb8526056e48e0846 : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/903426a05b1b4170bd0dcbddb791f429 : SUCCESS in 4m 47s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/aecae8262a7842499c34ce666474f533 : SUCCESS in 21m 18s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/0c85165089c84ef9a1d2e05a43267742 : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":7},{"id":"1d794554c004ef2d62cc55d5399644dd40726783","tag":"autogenerated:zuul:manual-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-12-01 02:12:47.000000000","message":"Patch Set 7:\n\n(9 comments)\n\nBuild succeeded (manual-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/3f28494e9c2f4536b9fb85322d82b318\n\n- teim-code-review https://zuul.teim.app/t/main/build/0c5d5966e03041b8ab813bb2cb4f21e7 : SUCCESS in 22m 45s\n\nWarning:\n  Comments left for invalid file watcher/common/executor.py","accounts_in_message":[],"_revision_number":7},{"id":"60b292d015eec636d47d6fafb3ce2fb5113b1d3d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-01 02:24:14.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/85c5430f0a694d5ba9f52e22f2ed3c82\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/72e700bbb0b34fb0a3459f8be5a6a0b5 : FAILURE in 32m 08s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5bcbd6ba31574d2c877e0a2dc29e2787 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":7},{"id":"0793d337d3ffcc0edfa2925b3db100f3578c4cc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 03:11:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/797fad0a1e9b4f519c607e85a1b1e4b4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/651ef07431eb46459f9aa621a3d05c9d : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8702df3d65b479c8c283a3b466dfe5c : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/464e6812e0e84f23a8f77743c9677885 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd43fdedf03c4d0e832c7e28be77cb02 : SUCCESS in 6m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ee8b4987ee8441097f0ea881b1b8957 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b902a3081a914b18bdf89919d8c2eb4a : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b10bef68210448f0b22bebe944c714d3 : SUCCESS in 4m 04s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/ab5027f94a62484e8be26ed2bf04e076 : SUCCESS in 5m 58s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/5ed5129e95674a3f9cedcbeb6dd3908d : SUCCESS in 29m 41s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/40d3de5e8b114510bf2b113ba9ea0288 : SUCCESS in 42m 54s\n- watcher-tempest-strategies https://zuul.opendev.org/t/openstack/build/06db2458012d408ba80d617217d0592f : SUCCESS in 1h 13m 10s\n- watcher-tempest-actuator https://zuul.opendev.org/t/openstack/build/8cfcea9b279549a883aae0b7780d4dcc : SUCCESS in 50m 27s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/b5451f1cb80149ec8990c6f63f30fdc0 : SUCCESS in 12m 38s\n- watcher-prometheus-integration https://zuul.opendev.org/t/openstack/build/70e3e52226dd4bdc99494dc53464b0e1 : SUCCESS in 1h 11m 25s\n- watcher-prometheus-integration-threading https://zuul.opendev.org/t/openstack/build/503e88b3678141b0bbeeaf25fcb2a69c : SUCCESS in 44m 36s\n- watcher-aetos-integration https://zuul.opendev.org/t/openstack/build/8f807d60bec5461ca6c6084b6bd8c470 : SUCCESS in 1h 15m 38s","accounts_in_message":[],"_revision_number":7},{"id":"e292301cc83b666655205281b0ee61f35addd8ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 03:12:25.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1a5a2c51975724e016c1514b8bc3d0c76832a6bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 03:46:24.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb6291e41ce14c9eaa426dbb1147ffad\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8dfe5be4fcd445eeb33588013800200f : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac969cff02f74bb49e6542fa76248006 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/708de5083ec743c199958e8213bc8b23 : SUCCESS in 6m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/954a8981e90f447187bdb6c43948863c : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/259054c01a7e43a7ac84f8d897c764a5 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f86bbe62b3be44cdaf9b585963d74cd0 : SUCCESS in 3m 46s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/39d14173ece74a93bc8fdd6124afe3da : SUCCESS in 29m 35s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/54c97fac91e146bf8f14ff34a23b3ee2 : SUCCESS in 15m 14s","accounts_in_message":[],"_revision_number":7},{"id":"b7e2f64b824c99c10e0ed0c3f02c98a94b058ad3","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-12-01 14:15:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"87b22c99b2c6f7d8e5d2badd0bb2eeb2013e2116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-12-29 13:12:41.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":8},{"id":"03091112a7979d2371546a0a5364e5bf20a8970e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-29 14:44:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a02e78dc35b140b5a7fbbf4bbeff3bdc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6aaca341c4d142cdbbb439454779e605 : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31aaf59abe844539b183c1d2e9909c97 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7d8274d581f4ffb9bba052b92a8c0f0 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cf3891a128b45e0a1d783738a27d851 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9643627c54e049a88db5864547e7b56c : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/742fa9686e1043caae460e4cd920e2d1 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec5a144467ce43109ae52643f9100309 : SUCCESS in 3m 40s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/bfae92156f1547c98a0d9c74dff6c6bf : SUCCESS in 3m 28s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/09aaf313a542447ab9ca191e4e0bcd52 : SUCCESS in 44m 08s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/f5a0ef572f16477f9437f8adb0daa0fc : SUCCESS in 34m 58s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/18f3041cf254481bb6ab75590af48bab : SUCCESS in 42m 57s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/c6b68be43320418185e1a6a67bfa2428 : SUCCESS in 1h 24m 50s\n- watcher-tempest-prometheus-threading https://zuul.opendev.org/t/openstack/build/58f8878684804edfbf896048ce431254 : SUCCESS in 44m 37s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/ed4c788276cd40bca76caf287403cc0e : SUCCESS in 1h 06m 55s","accounts_in_message":[],"_revision_number":8},{"id":"9fc998928235c64cc47d7769cecaa5ae97cfee3b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-29 15:49:57.000000000","message":"Patch Set 8: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/1b70d2dc4fcc482f81804fd81186bd51\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/58b56e1821ff400f9886db41a0098f11 : SUCCESS in 2h 35m 53s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/3267956ba4ac47879ca35068949353ea : SUCCESS in 1h 57m 38s","accounts_in_message":[],"_revision_number":8},{"id":"bb1996405d1469bb3c070bcc3698a14602e2a7ef","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-12-30 15:44:14.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c3921bf2825c62dfbfa5102d41e4de1907451e96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 15:44:56.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"d33a8f18eb21b115ca0314d7fff02dc82a8cd626","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 17:05:27.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffb6b2f90a2d40b6b7c123359812aad4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2868dc4c8afd45c193d5fbdba5b5004b : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1597ad630a4b038a449d688cde8360 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39f6f00f175a4501a87950695bc60dc3 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b68182793a1749f38576ed371673e5bb : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6d098f8ae54671b3dde7b65e7ac901 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bc8e50bde304302bfa570156a2c4bdd : SUCCESS in 3m 50s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/dbd70d4cf6254571942da37a101bb953 : SUCCESS in 4m 54s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/ffc96d8b39484759b26ee4f037d9e8e2 : SUCCESS in 44m 38s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/2354cc3c87054283b369912ad2691a29 : SUCCESS in 36m 38s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/f6b6ad072f5d4a5f859c711cf97ea69e : SUCCESS in 58m 16s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/01ab72dc01384000967c47ddae547d12 : SUCCESS in 1h 16m 05s\n- watcher-tempest-prometheus-threading https://zuul.opendev.org/t/openstack/build/7f39083159ad44c7aece16b10131842a : SUCCESS in 47m 18s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/5ac2ae7acbc04cc185a54145b7f6c8ff : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":8},{"id":"b33427073e2cecf7924ffc86172ad878fa3789d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 17:05:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"468104b085b9abef099684dfdc0c475233a15d3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 17:07:03.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cb08501737846159a2451aac3bca788\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d3deab8946d04054bb3a11d693369e14 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cab4d3287384240bb80d33f404b64d8 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78f3ccf369d64c02860a01a6ff621595 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e45860787f9213c4866b60d6ef7e787bc97c0fdc","revisions":{"14e58c28eb4c5a1f2db2f8c4c57e1553ce0a37e8":{"kind":"REWORK","_number":1,"created":"2025-11-05 20:46:32.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/26/966226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/26/966226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/26/966226/1"}}},"commit":{"parents":[{"commit":"b5725d6ea60d3b7fb2d2b808b261ccdc547df7c4","subject":"Add second instance of watcher-decision-engine in the compute node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/b5725d6ea60d3b7fb2d2b808b261ccdc547df7c4"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-10-09 19:34:05.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-11-05 20:46:02.000000000","tz":-180},"subject":"WIP - Adds support for threading mode in applier","message":"WIP - Adds support for threading mode in applier\n\nChange-Id: I85926cdd1e4393be42fe6d543af50bf2948b5ce6\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/14e58c28eb4c5a1f2db2f8c4c57e1553ce0a37e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/14e58c28eb4c5a1f2db2f8c4c57e1553ce0a37e8"}]},"branch":"refs/heads/master"},"e6d1d1d5738320781591401a7fa7b85e85fcbbb4":{"kind":"REWORK","_number":2,"created":"2025-11-12 13:47:54.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/26/966226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/26/966226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/26/966226/2"}}},"commit":{"parents":[{"commit":"b5725d6ea60d3b7fb2d2b808b261ccdc547df7c4","subject":"Add second instance of watcher-decision-engine in the compute node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/b5725d6ea60d3b7fb2d2b808b261ccdc547df7c4"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-10-09 19:34:05.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-11-12 13:44:21.000000000","tz":-180},"subject":"Adds support for threading mode in applier","message":"Adds support for threading mode in applier\n\nTo support threading mode in the applier, this patch switches taskflow\nengine to serial when eventlet is not patched. This avoids an issue\ncaused by the multiple threads reading/writing Action objects when\nrunning with parallel engine. This limitation should be adressed in\na different patch which may also require a refactoring in database api\nand how watcher manages its sessions and contexts.\n\nChanges:\n1. Add InlineThread class\n   - Executes functions synchronously while maintaining thread interface\n2. Use serial engine for threading mode\n   - Switches to taskflow serial engine when eventlet is not patched\n3. Replace default GreenThreadPoolExecutor in TriggerActionPlan\n   - A new executor will be created based on the thread mode configured\n4. Adds helper functions to spawn, start, wait and kill threads\n   - The thread will be handled based on the thread mode configured\n5. Threading mode job and tox-threading updated to include applier\n   - These jobs will now enable threading mode for the applier content\n     and service\n\nAssisted-By: claude-code sonnet-4.5\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I85926cdd1e4393be42fe6d543af50bf2948b5ce6\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/e6d1d1d5738320781591401a7fa7b85e85fcbbb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/e6d1d1d5738320781591401a7fa7b85e85fcbbb4"}]},"branch":"refs/heads/master"},"ace8e0f719f06553ac013e8bf996d876daceb55b":{"kind":"REWORK","_number":3,"created":"2025-11-12 18:04:08.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/26/966226/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/26/966226/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/26/966226/3"}}},"commit":{"parents":[{"commit":"b5725d6ea60d3b7fb2d2b808b261ccdc547df7c4","subject":"Add second instance of watcher-decision-engine in the compute node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/b5725d6ea60d3b7fb2d2b808b261ccdc547df7c4"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-10-09 19:34:05.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-11-12 18:00:09.000000000","tz":-180},"subject":"Adds support for threading mode in applier","message":"Adds support for threading mode in applier\n\nTo support threading mode in the applier, this patch switches taskflow\nengine to serial when eventlet is not patched. This avoids an issue\ncaused by the multiple threads reading/writing Action objects when\nrunning with parallel engine. This limitation should be adressed in\na different patch which may also require a refactoring in database api\nand how watcher manages its sessions and contexts.\n\nChanges:\n1. Add InlineThread class\n   - Executes functions synchronously while maintaining thread interface\n2. Use serial engine for threading mode\n   - Switches to taskflow serial engine when eventlet is not patched\n3. Replace default GreenThreadPoolExecutor in TriggerActionPlan\n   - A new executor will be created based on the thread mode configured\n4. Adds helper functions to spawn, start, wait and kill threads\n   - The thread will be handled based on the thread mode configured\n5. Threading mode job and tox-threading updated to include applier\n   - These jobs will now enable threading mode for the applier content\n     and service\n\nAssisted-By: claude-code sonnet-4.5\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I85926cdd1e4393be42fe6d543af50bf2948b5ce6\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/ace8e0f719f06553ac013e8bf996d876daceb55b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/ace8e0f719f06553ac013e8bf996d876daceb55b"}]},"branch":"refs/heads/master"},"59149edc2d80979d984e0d463ba628d566a94f89":{"kind":"REWORK","_number":4,"created":"2025-11-13 17:29:00.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/26/966226/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/26/966226/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/26/966226/4"}}},"commit":{"parents":[{"commit":"b5725d6ea60d3b7fb2d2b808b261ccdc547df7c4","subject":"Add second instance of watcher-decision-engine in the compute node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/b5725d6ea60d3b7fb2d2b808b261ccdc547df7c4"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-10-09 19:34:05.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-11-13 17:27:07.000000000","tz":-180},"subject":"Adds support for threading mode in applier","message":"Adds support for threading mode in applier\n\nTo support threading mode in the applier, this patch switches taskflow\nengine to serial when eventlet is not patched. This avoids an issue\ncaused by the multiple threads reading/writing Action objects when\nrunning with parallel engine. This limitation should be adressed in\na different patch which may also require a refactoring in database api\nand how watcher manages its sessions and contexts.\n\nChanges:\n1. Add InlineThread class\n   - Executes functions synchronously while maintaining thread interface\n2. Use serial engine for threading mode\n   - Switches to taskflow serial engine when eventlet is not patched\n3. Replace default GreenThreadPoolExecutor in TriggerActionPlan\n   - A new executor will be created based on the thread mode configured\n4. Adds helper functions to spawn, start, wait and kill threads\n   - The thread will be handled based on the thread mode configured\n5. Threading mode job and tox-threading updated to include applier\n   - These jobs will now enable threading mode for the applier content\n     and service\n\nAssisted-By: claude-code sonnet-4.5\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I85926cdd1e4393be42fe6d543af50bf2948b5ce6\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/59149edc2d80979d984e0d463ba628d566a94f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/59149edc2d80979d984e0d463ba628d566a94f89"}]},"branch":"refs/heads/master"},"fb9225727f45b72426b459d769fc8d62d530b179":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-11-17 15:21:08.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/26/966226/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/26/966226/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/26/966226/5"}}},"commit":{"parents":[{"commit":"4045158e1d325b052ef09314caf84d8ac8d25eb6","subject":"[DB] adopt the AdHocDbFixture from oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/4045158e1d325b052ef09314caf84d8ac8d25eb6"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-10-09 19:34:05.000000000","tz":-180},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-17 15:20:55.000000000","tz":0},"subject":"Adds support for threading mode in applier","message":"Adds support for threading mode in applier\n\nTo support threading mode in the applier, this patch switches taskflow\nengine to serial when eventlet is not patched. This avoids an issue\ncaused by the multiple threads reading/writing Action objects when\nrunning with parallel engine. This limitation should be adressed in\na different patch which may also require a refactoring in database api\nand how watcher manages its sessions and contexts.\n\nChanges:\n1. Add InlineThread class\n   - Executes functions synchronously while maintaining thread interface\n2. Use serial engine for threading mode\n   - Switches to taskflow serial engine when eventlet is not patched\n3. Replace default GreenThreadPoolExecutor in TriggerActionPlan\n   - A new executor will be created based on the thread mode configured\n4. Adds helper functions to spawn, start, wait and kill threads\n   - The thread will be handled based on the thread mode configured\n5. Threading mode job and tox-threading updated to include applier\n   - These jobs will now enable threading mode for the applier content\n     and service\n\nAssisted-By: claude-code sonnet-4.5\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I85926cdd1e4393be42fe6d543af50bf2948b5ce6\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/fb9225727f45b72426b459d769fc8d62d530b179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/fb9225727f45b72426b459d769fc8d62d530b179"}]},"branch":"refs/heads/master"},"480199511046aae5085e7fe9cc1dd616f83dc740":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-11-17 16:47:34.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/26/966226/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/26/966226/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/26/966226/6"}}},"commit":{"parents":[{"commit":"68bc99ab71aa724664eaf0be507df804c9496649","subject":"[DB] adopt the AdHocDbFixture from oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/68bc99ab71aa724664eaf0be507df804c9496649"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-10-09 19:34:05.000000000","tz":-180},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-17 16:47:17.000000000","tz":0},"subject":"Adds support for threading mode in applier","message":"Adds support for threading mode in applier\n\nTo support threading mode in the applier, this patch switches taskflow\nengine to serial when eventlet is not patched. This avoids an issue\ncaused by the multiple threads reading/writing Action objects when\nrunning with parallel engine. This limitation should be adressed in\na different patch which may also require a refactoring in database api\nand how watcher manages its sessions and contexts.\n\nChanges:\n1. Add InlineThread class\n   - Executes functions synchronously while maintaining thread interface\n2. Use serial engine for threading mode\n   - Switches to taskflow serial engine when eventlet is not patched\n3. Replace default GreenThreadPoolExecutor in TriggerActionPlan\n   - A new executor will be created based on the thread mode configured\n4. Adds helper functions to spawn, start, wait and kill threads\n   - The thread will be handled based on the thread mode configured\n5. Threading mode job and tox-threading updated to include applier\n   - These jobs will now enable threading mode for the applier content\n     and service\n\nAssisted-By: claude-code sonnet-4.5\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I85926cdd1e4393be42fe6d543af50bf2948b5ce6\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/480199511046aae5085e7fe9cc1dd616f83dc740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/480199511046aae5085e7fe9cc1dd616f83dc740"}]},"branch":"refs/heads/master"},"6f75b90fd9c0edf195d1963c209b076575f0822a":{"kind":"REWORK","_number":7,"created":"2025-11-19 20:29:25.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/26/966226/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/26/966226/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/26/966226/7"}}},"commit":{"parents":[{"commit":"68bc99ab71aa724664eaf0be507df804c9496649","subject":"[DB] adopt the AdHocDbFixture from oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/68bc99ab71aa724664eaf0be507df804c9496649"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-10-09 19:34:05.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-11-19 20:29:06.000000000","tz":-180},"subject":"Adds support for threading mode in applier","message":"Adds support for threading mode in applier\n\nTo support threading mode in the applier, this patch switches taskflow\nengine to serial when eventlet is not patched. This avoids an issue\ncaused by the multiple threads reading/writing Action objects when\nrunning with parallel engine. This limitation should be adressed in\na different patch which may also require a refactoring in database api\nand how watcher manages its sessions and contexts.\n\nChanges:\n1. Add InlineThread class\n   - Executes functions synchronously while maintaining thread interface\n2. Use serial engine for threading mode\n   - Switches to taskflow serial engine when eventlet is not patched\n3. Replace default GreenThreadPoolExecutor in TriggerActionPlan\n   - A new executor will be created based on the thread mode configured\n4. Adds helper functions to spawn, start, wait and kill threads\n   - The thread will be handled based on the thread mode configured\n5. Threading mode job and tox-threading updated to include applier\n   - These jobs will now enable threading mode for the applier content\n     and service\n\nAssisted-By: claude-code sonnet-4.5\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I85926cdd1e4393be42fe6d543af50bf2948b5ce6\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/6f75b90fd9c0edf195d1963c209b076575f0822a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/6f75b90fd9c0edf195d1963c209b076575f0822a"}]},"branch":"refs/heads/master"},"e45860787f9213c4866b60d6ef7e787bc97c0fdc":{"kind":"REWORK","_number":8,"created":"2025-12-29 13:12:41.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/26/966226/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/26/966226/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/26/966226/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/26/966226/8"}}},"commit":{"parents":[{"commit":"1d32e734f34db23b5aa444f9c3d22a13bef058b1","subject":"Retry calls to nova API when connection errors are detected","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/1d32e734f34db23b5aa444f9c3d22a13bef058b1"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-10-09 19:34:05.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-12-29 13:12:05.000000000","tz":-180},"subject":"Adds support for threading mode in applier","message":"Adds support for threading mode in applier\n\nTo support threading mode in the applier, this patch switches taskflow\nengine to serial when eventlet is not patched. This avoids an issue\ncaused by the multiple threads reading/writing Action objects when\nrunning with parallel engine. This limitation should be adressed in\na different patch which may also require a refactoring in database api\nand how watcher manages its sessions and contexts.\n\nChanges:\n1. Add InlineThread class\n   - Executes functions synchronously while maintaining thread interface\n2. Use serial engine for threading mode\n   - Switches to taskflow serial engine when eventlet is not patched\n3. Replace default GreenThreadPoolExecutor in TriggerActionPlan\n   - A new executor will be created based on the thread mode configured\n4. Adds helper functions to spawn, start, wait and kill threads\n   - The thread will be handled based on the thread mode configured\n5. Threading mode job and tox-threading updated to include applier\n   - These jobs will now enable threading mode for the applier content\n     and service\n\nAssisted-By: claude-code sonnet-4.5\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I85926cdd1e4393be42fe6d543af50bf2948b5ce6\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/e45860787f9213c4866b60d6ef7e787bc97c0fdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/e45860787f9213c4866b60d6ef7e787bc97c0fdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
