)]}'
{"id":"openstack%2Fwatcher~999477","triplet_id":"openstack%2Fwatcher~master~If190562129c5d2d191f9ac86f3cd67305af154d3","project":"openstack/watcher","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-24 07:35:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If190562129c5d2d191f9ac86f3cd67305af154d3","subject":"zuul: Drop redundant override of cinder option","status":"MERGED","created":"2026-08-01 01:01:00.000000000","updated":"2026-09-24 07:37:01.000000000","submitted":"2026-09-24 07:35:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999477","meta_rev_id":"7f70ea9efe63a0c58f17392f3a098531ca1317e1","_number":999477,"virtual_id_number":999477,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-24 07:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"all":[{"value":2,"date":"2026-09-10 09:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":-1,"date":"2026-08-24 06:14:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-09-23 15:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-09-23 15:20:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}],"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-01 01:01:06.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-08-01 01:11:58.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-01 02:55:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 09:29:20.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"REVIEWER"},{"updated":"2026-09-23 15:20:16.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"bfdabce0b3d450aa9587633a2a24e9d534234e3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-01 01:01:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27fc7f04a4e9ba6c5124e7209134ea74f8a4456d","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-01 01:01:06.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e41a07c7def85585ee2ead2121b53e9201df95e","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-01 01:06:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/3dc88bd1cb62442fb7d4071cdbd1ee2b\n\n- teim-code-review https://zuul.teim.app/t/main/build/c1c083c3f00b4bb787c6f3482ced8581 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"741a568a08b887170a8783b4e45c9e89cd0717fb","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-01 01:11:58.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/7cbd8f36fcd545d3a92f01ca479dbdc6\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/84a498ee379f429d8ce20e48d26818ff : FAILURE in 8m 58s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/2ed64757439d4012babf889ceb523d2a : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":1},{"id":"a3a3f93f7dca1674c587b7e3224894681c30a36d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-01 02:55:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56e5c15a430c45a5ba3d6860eb83342f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82f333a4c9944894ac3585fc8efc0161 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b63ce8c795247828da44196a6c10306 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b818b70fa23d45058ff57e5d2b7e90cc : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fefaea93d8f4fd98392e1df00768e5d : SUCCESS in 6m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b6ddc07af5b94653a7f0886920ea0da0 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71a089bba78b451db8f61a709f1d4117 : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69e8eb2b6ee6489e93d28e7cf03130ca : SUCCESS in 5m 51s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/beea515d708d44509ae5a8cc408e303a : SUCCESS in 3m 23s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/64722ea04152420eac956ff9d3175b27 : SUCCESS in 53m 31s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9e5b0b52889142fca5b6e8d286268a44 : SUCCESS in 52m 41s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/d985468a8dd94868827e4c82771b1862 : SUCCESS in 29m 49s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/76f1b19e7bf440c898969d8724c681f3 : SUCCESS in 1h 39m 56s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/796d2c58e68c48c296ec3efbd9475603 : SUCCESS in 1h 34m 39s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/7d2a35e63bdd45b5acefdd3f5de2690a : SUCCESS in 1h 34m 35s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/64b84a6e11c2421782ac89c0410a932b : FAILURE in 1h 47m 25s","accounts_in_message":[],"_revision_number":1},{"id":"41406df57d5dd2268bc9ab4f5f9aaa5510b2829c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-24 06:03:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"87914577feaf77e3aaee7313dd03556193297049","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-24 06:04:03.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b2efa0694b6a2692a651e499ddb71c474908010","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-24 06:14:35.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/09e66dc882474b53801570e294cde3b3\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/d909936296be4d7c9a03c2efa7721439 : FAILURE in 8m 48s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/80f4c409e14c4403a17361a07b49dacb : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":2},{"id":"3d91eb2b97cad7819650ed5914c52b6c75b32432","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-24 07:11:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/d9a3606acae24b6cb9d6769bd9b2d1c2\n\n- teim-code-review https://zuul.teim.app/t/main/build/7d629c38d2d64fa6a51aeb5ae9f2b60a : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":2},{"id":"4c1391db59d7d413af7adecf074c0645ead021b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 07:47:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce800417953543408688161e94402801\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee72282bf461472591c251c15f638b4e : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f43c988d9e4cffbd5dfb4b4b0f21e3 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/672c6155a3804d22b7787bf4411a73d0 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fab222cb028d44a88666502e6cc1c569 : SUCCESS in 4m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1f3a3ab6ee24984a3d181cadae31644 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f833eaf2777b4d2ca40210744764c038 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed3a51011bc3429b9282af4437796025 : SUCCESS in 4m 28s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/d2450d07d44346baa6841d6c65b14ed8 : SUCCESS in 6m 37s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/5d36e36f6e63419e818e8d8416f0891e : SUCCESS in 25m 49s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c3124a7b0fa4db8b375b07c761e380c : SUCCESS in 52m 43s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/0f1ae217f53f42a9a35b596549c038a5 : SUCCESS in 41m 38s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/c9691d92dadd4e65bd3ce437c219db1d : SUCCESS in 1h 02m 33s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/764e7f7a2b754e9296061f0d6b2e9062 : SUCCESS in 1h 35m 24s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/05418b1bcd0c4b2e96246f5fdab16a86 : SUCCESS in 1h 24m 58s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/df548bcda41f4f58b2fcbf102165f875 : SUCCESS in 1h 33m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/37ec01696da54ea7b91c8deb9a3bf27e : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f262504418b5bced4006192379f0862120bb16f3","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-09-10 09:29:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7aa0e9fdc688baa7b9e2f37e52cef9b68ae9644","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-09-23 15:20:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"83dfb3862d367f1cda67491f04e163b3e664fdd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 15:21:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8fbaff02594e6991b954b0bbf4b23b259d997086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 17:15:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f5751c9ac874cf8983dedbbbadd2575\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3568e086344561b53296b60ac69521 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd83758d98c64e5885a1d28b2e33846f : SUCCESS in 5m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a1cafa6290584b55bbee80d35684eee4 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f184c5f0aa47e5917fc3e7668509b3 : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f65dea597c1a4c97a456b9ed9d3731ae : SUCCESS in 4m 53s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/ae6324a0502b4bdf83294c2d3670bdf9 : SUCCESS in 57m 36s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/259041aa06ba42f3a96df614e1dc9b8c : SUCCESS in 1h 00m 54s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/5ab194e9a17944158ca42a40a2c57806 : SUCCESS in 27m 24s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/771abaa07df94eb2b08a3f68aacb7adf : FAILURE in 1h 27m 06s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/7fe3647cf78d4efe8d777007aac15291 : SUCCESS in 52m 46s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/bd549e0eb13a403ebff989fc8aa0fe81 : SUCCESS in 1h 47m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fd2519043dfd48a897d9ecee8b4c49f2 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"8526e76d14581868056639849e28e988c12eb52a","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-09-23 17:39:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"797ea7016e9dc9da8e463bec5b563185374325c5","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-23 17:39:20.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"70d3506e2526991e5fbbe5c21f666380621c9512","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-23 17:48:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/8c196b2fd4794f8987cb1331ec04e3b4\n\n- teim-code-review https://zuul.teim.app/t/main/build/3066e9e6984f4dcbb0757ad9d29d4a85 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"5b67612677127d5a7219c654bbcf50cac84a88ea","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-23 18:59:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/0f36b1e234574ddd808900eb25b4a2ee\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/97f94019ecd0453bad5badc76243e7d7 : FAILURE in 8m 35s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/81579034b3454f328fe752248201ce31 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":2},{"id":"8d2050ae7cbe98bea40821a365375dc4f8aa6ace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 19:25:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e74a4b2b4ce549008351b5de1d4afb60\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d823d8176f941678d397430bf086a5d : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2200bb2cb6bc43eb97630fce9a9e5973 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4128c137918547aaae958a21d90a6b87 : SUCCESS in 5m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b5f075cec88d48f294ab9b1f8253fcd3 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d65383b834f45d98187a955830b015d : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a339adc7a89f487f9839264e5f5aaaa9 : SUCCESS in 4m 59s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/50c8b4899b18437cb673e2a43c9b653a : SUCCESS in 25m 25s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b32735bbf25942a8af171049abc461fa : SUCCESS in 51m 35s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/262c570963dc467f8f3b60f8d4f0e9a6 : SUCCESS in 38m 26s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/3cabb436312540abbdc97d4da99bf97b : SUCCESS in 1h 38m 53s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/2431765dc0f142e8a6e09a708e2d497f : SUCCESS in 58m 26s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/69e80b7beb61419cb845c49e0bdc6953 : FAILURE in 1h 34m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c49832b4a3d144529d7df49901f8ad9e : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":2},{"id":"b571d87fbfe321aa8b2a1599c5f39354f18056a3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-24 03:56:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2dbff0021e31af343c48ceafec5644efd6957b67","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-24 03:56:35.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b31f9ed03098382d2afb91a736d957b9d385acf9","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-24 04:00:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/4f461a123c0c4bd0b285b3cf74f48bc5\n\n- teim-code-review https://zuul.teim.app/t/main/build/1d6cc8b656064c08857f01421d2d985a : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b9ccfa378af1af1136cdd39e83c49fefb98f19a0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-24 04:06:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/092f652243a44ea2b82cb1abdadb4b8b\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/65d32434b572456aadc534496b3ca721 : FAILURE in 8m 39s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/6e73b9a14ca642e9a6cf26b87e12323a : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":2},{"id":"ec6ce6a4515a29fcd1f94eba0c379ceba42fdf66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 05:44:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cec58e55dc4a4710ab6d31c46348c671\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70eddf02727941ebb8d0f771e47ca1e5 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f7d93222eda4e4ba345472914a68d21 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45dade61568144cc8709f941d72cbca0 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1dab8aad95bb4595bdf01f9618fbee25 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f86e206fd99e4764998f368d978a87de : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92123fe495cc4d9c850b97c75f88be1c : SUCCESS in 3m 31s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/260ab466a45246adb6a2bdf4c08c87c4 : SUCCESS in 40m 35s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/58734106db99463b84bb3d45634f01ac : SUCCESS in 38m 20s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/7a3d7799853947ea9a50fc03edcc064a : SUCCESS in 16m 59s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/92cb112b020c47ab80138f000826dbe4 : SUCCESS in 1h 38m 43s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/18035ad7b0bd49ad97153ed8b75fe43c : SUCCESS in 1h 43m 50s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/47bd56ee0c3f4172ada75884ee132fd0 : SUCCESS in 1h 08m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fb64843aa7b34d17ab1c4e24c3ac85ae : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"210eec4159daa72ab388c0ee4b38ae3c4f69a6ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 05:47:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ca19c0321fed4611c56bfd654e146f37fb2ef78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 07:35:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d008286f79d9451498b4c23fedfce92f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497fef95e33341cc97c6a44efd10f210 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eeb9fda6738641aa9850e73460cff2d1 : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e10c27b696fe440bb3f73e8891f74bd5 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278fbe02046d44fd9df57ab706c68de1 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ef6062f56e142d2ad56c1938995b1b5 : SUCCESS in 6m 41s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/82d986e3018e4ffbaa3f33715b5e35ac : SUCCESS in 52m 44s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b261f72e00b0472ca818956344de243a : SUCCESS in 49m 50s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/dfde56b1f8be43a7ae0273744c62fbe2 : SUCCESS in 41m 21s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/47bb5185bc7744f5a22c851c23a4fc38 : SUCCESS in 1h 40m 28s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/36cd5984c51f46599aaf7112ee53c974 : SUCCESS in 49m 34s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/9fc36e52dcea469194b206583b05e7e1 : SUCCESS in 1h 44m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6ebf9e8430c145278df8ee5d492ab6e2 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"4c33c5f9dd58db327dd25a9df7a6fdfd79e698f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 07:35:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7f70ea9efe63a0c58f17392f3a098531ca1317e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 07:37:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea808be1c08f49b787d97d49291d842d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59db8c0d4afb44049c0548a84b0211d5 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ddb66f01be39e5646e7805ece089d25892ea8284","revisions":{"1972ed87719c3de650f3123754f87752fd9ab02e":{"kind":"REWORK","_number":1,"created":"2026-08-01 01:01:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/999477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/77/999477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/77/999477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/77/999477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/77/999477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/77/999477/1"}}},"commit":{"parents":[{"commit":"4bca4144c8fadd2fbcb4720253488b35e94bbeba","subject":"Merge \"Replace python-cinderclient with openstacksdk\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/4bca4144c8fadd2fbcb4720253488b35e94bbeba"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-01 01:00:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-01 01:00:15.000000000","tz":540},"subject":"zuul: Drop redundant override of cinder option","message":"zuul: Drop redundant override of cinder option\n\nThe [oslo_messaging_notifications] driver option is already set by\ndevstack when ceilometer is enabled.\n\nChange-Id: If190562129c5d2d191f9ac86f3cd67305af154d3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/1972ed87719c3de650f3123754f87752fd9ab02e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/1972ed87719c3de650f3123754f87752fd9ab02e"}]},"branch":"refs/heads/master"},"ddb66f01be39e5646e7805ece089d25892ea8284":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-24 06:03:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/999477/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/77/999477/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/77/999477/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/77/999477/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/77/999477/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/77/999477/2"}}},"commit":{"parents":[{"commit":"ec06744bffd4ba858da7f4996cc872723dfbde70","subject":"Merge \"Add Nova/Placement emulators and host_maintenance functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/ec06744bffd4ba858da7f4996cc872723dfbde70"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-01 01:00:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 06:03:55.000000000","tz":0},"subject":"zuul: Drop redundant override of cinder option","message":"zuul: Drop redundant override of cinder option\n\nThe [oslo_messaging_notifications] driver option is already set by\ndevstack when ceilometer is enabled.\n\nChange-Id: If190562129c5d2d191f9ac86f3cd67305af154d3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/ddb66f01be39e5646e7805ece089d25892ea8284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/ddb66f01be39e5646e7805ece089d25892ea8284"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"4bca4144c8fadd2fbcb4720253488b35e94bbeba","ours":"1972ed87719c3de650f3123754f87752fd9ab02e","theirs":"ec06744bffd4ba858da7f4996cc872723dfbde70","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
