)]}'
{"id":"openstack%2Fwhereto~580738","triplet_id":"openstack%2Fwhereto~master~I27dcb4fbb58fc366e213dac7bad9bb56235568cc","project":"openstack/whereto","branch":"master","topic":"python3-first","hashtags":[],"change_id":"I27dcb4fbb58fc366e213dac7bad9bb56235568cc","subject":"switch to PTI documentation jobs","status":"MERGED","created":"2018-07-06 18:37:20.000000000","updated":"2018-07-10 15:13:27.000000000","submitted":"2018-07-10 15:13:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"580738-1531235607248-0bf9a1e8","meta_rev_id":"adc6796c4a1713354129a89bb85dad349561b224","_number":580738,"virtual_id_number":580738,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2018-07-10 14:58:43.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2018-07-10 15:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2018-07-09 18:01:46.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2018-07-10 15:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2018-07-10 15:02:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-09 18:01:46.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2018-07-10 15:02:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2018-07-10 15:13:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74e743ccc8717f136fe844626514e55a83829db6","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-07-06 18:37:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32ff83d64725172f184780b61d70d146c84b7f2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-06 18:37:47.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/whereto on branch master.  The error was:\n\n  The project template \"publish-openstack-sphinx-docs-pti\" was not\n  found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs\n        - openstack-python35-jobs\n        - publish-openstack-sphinx-docs-pti\n\n  in \"openstack/whereto/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"70bb7c7af3ff27ac306cfdc90e80a47afc5fdb56","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-07-06 18:58:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef924381c730b1a7ca93f609e2366322a2bb1b33","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-07-09 16:08:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"019f441f854a43e96a071a895f761146fb481aed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-09 16:08:28.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/whereto on branch master.  The error was:\n\n  The project template \"publish-openstack-sphinx-docs-pti\" was not\n  found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs\n        - openstack-python35-jobs\n        - publish-openstack-sphinx-docs-pti\n\n  in \"openstack/whereto/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"3383f4e5837feaef367f3b5a2f22fa97e19f9229","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-07-09 17:51:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b14c59dbd5aa45e0d9e0076834a8fd524d89d83f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-09 17:58:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/38/580738/2/check/openstack-tox-pep8/198b7a7/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/38/580738/2/check/openstack-tox-py27/6a038dd/ : SUCCESS in 2m 52s\n- openstack-tox-py35 http://logs.openstack.org/38/580738/2/check/openstack-tox-py35/1f7acbd/ : SUCCESS in 2m 41s\n- openstack-tox-docs http://logs.openstack.org/38/580738/2/check/openstack-tox-docs/55e0313/ : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"cba73b8ca5cc7350e9685a22d9d37e394f38234f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-07-09 18:01:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6aaa7e557ff83f6f534cf684a4bd0f2656b7823c","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-07-10 14:58:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5646e5976f38fd460f73196882c11351abf98f5f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-07-10 15:02:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"63b8e0194da29ec8d94645e265782ecf579f6ce5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 15:03:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"db88a5e3a32ba3d6a980e3a2f7e5bb9b4c636b11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 15:13:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/38/580738/2/gate/openstack-tox-pep8/5e987b6/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/38/580738/2/gate/openstack-tox-py27/27c0735/ : SUCCESS in 4m 10s\n- openstack-tox-py35 http://logs.openstack.org/38/580738/2/gate/openstack-tox-py35/9ac809c/ : SUCCESS in 3m 12s\n- openstack-tox-docs http://logs.openstack.org/38/580738/2/gate/openstack-tox-docs/d1a89ef/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"1122567fe3a5995b1a855956d05233175f4887c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 15:13:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b1292a23f190461cf6e60961aa3adb23a7a973d5","revisions":{"a6221f558ac9bfc1e0fc326c09711bb7891ac97f":{"kind":"REWORK","_number":1,"created":"2018-07-06 18:37:20.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/38/580738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/38/580738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/580738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/580738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/580738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/38/580738/1"}}},"commit":{"parents":[{"commit":"b30c8e7760c633c50a9fab7fb3294ee4265cb579","subject":"import zuul settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/b30c8e7760c633c50a9fab7fb3294ee4265cb579"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-07-06 18:35:22.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-07-06 18:36:54.000000000","tz":-240},"subject":"switch to PTI documentation jobs","message":"switch to PTI documentation jobs\n\nDepends-On: https://review.openstack.org/580485\nChange-Id: I27dcb4fbb58fc366e213dac7bad9bb56235568cc\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/a6221f558ac9bfc1e0fc326c09711bb7891ac97f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/a6221f558ac9bfc1e0fc326c09711bb7891ac97f"}]},"branch":"refs/heads/master"},"b1292a23f190461cf6e60961aa3adb23a7a973d5":{"kind":"REWORK","_number":2,"created":"2018-07-09 17:51:06.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/38/580738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/38/580738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/580738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/580738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/580738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/38/580738/2"}}},"commit":{"parents":[{"commit":"b30c8e7760c633c50a9fab7fb3294ee4265cb579","subject":"import zuul settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/b30c8e7760c633c50a9fab7fb3294ee4265cb579"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-07-06 18:35:22.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-07-09 17:50:58.000000000","tz":-240},"subject":"switch to PTI documentation jobs","message":"switch to PTI documentation jobs\n\nDepends-On: https://review.openstack.org/580485\nChange-Id: I27dcb4fbb58fc366e213dac7bad9bb56235568cc\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/b1292a23f190461cf6e60961aa3adb23a7a973d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/b1292a23f190461cf6e60961aa3adb23a7a973d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
