)]}'
{"id":"openstack%2Fwhereto~580741","triplet_id":"openstack%2Fwhereto~master~I1e0aeae7b0cbc44f4b5d2054fec91ea5552fda22","project":"openstack/whereto","branch":"master","topic":"python3-first","hashtags":[],"change_id":"I1e0aeae7b0cbc44f4b5d2054fec91ea5552fda22","subject":"fix bug link in readme","status":"MERGED","created":"2018-07-06 18:53:29.000000000","updated":"2018-07-10 15:13:27.000000000","submitted":"2018-07-10 15:13:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"580741-1531235607952-01c0f2cd","meta_rev_id":"c6b06a9aa664932ea735d882f1aacecd9ead277f","_number":580741,"virtual_id_number":580741,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2018-07-10 15:00:28.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2018-07-10 15:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2018-07-09 17:51:06.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2018-07-10 15:05:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2018-07-10 15:05:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-06 18:58:18.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2018-07-10 15:05:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2018-07-10 15:13:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b2dd532b6b00973543d9496a713454badcb8afe","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-07-06 18:53:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8246eeb17e1e176b96c347c6e18d61d7e4cddaff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-06 18:54:03.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/whereto on branch master.  The error was:\n\n  The project template \"publish-openstack-sphinx-docs-pti\" was not\n  found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs\n        - openstack-python35-jobs\n        - publish-openstack-sphinx-docs-pti\n\n  in \"openstack/whereto/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"980dd908080dde24539d6b9ade4a671fd85b6f24","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-07-06 18:58:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12366a8e26f84fc9b383ce809fecbbee7fb5a423","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-07-09 16:08:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"184b5a2eedcf9966305d54c8e1b6e39c57a0c82c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-09 16:09:01.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/whereto on branch master.  The error was:\n\n  The project template \"publish-openstack-sphinx-docs-pti\" was not\n  found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs\n        - openstack-python35-jobs\n        - publish-openstack-sphinx-docs-pti\n\n  in \"openstack/whereto/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"c17d18f5334ce266f591c39eeb9dcc275353af20","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-07-09 17:51:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b57c8a31d8b6aab6339c5a1391e289160e0f6fff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-09 18:00:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/580741/2/check/openstack-tox-pep8/a000935/ : SUCCESS in 3m 07s\n- openstack-tox-docs http://logs.openstack.org/41/580741/2/check/openstack-tox-docs/388ebe9/ : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":2},{"id":"0da82ea84904f96cdf7cb45c1fc9e37804efda5d","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-07-10 15:00:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"31710f41da98b409af88728661b01b72527be949","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-07-10 15:05:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nYup, this is running the correct target so looks good to me","accounts_in_message":[],"_revision_number":2},{"id":"305f2e42f8bbe80bffe489791a00a0235de84cfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 15:05:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a22d8583e87a4856dd9d866f5925085b03a53fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 15:13:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"227fc15925946a682c892d7c83de716c975ce788","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 15:13:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/580741/2/gate/openstack-tox-pep8/d1c4289/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/41/580741/2/gate/openstack-tox-docs/f3c5724/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e9788f4f48c09d9e5869e4279110c88af58c4216","revisions":{"288487af7c7c58c439efb22fec5ba6d1fbc47feb":{"kind":"REWORK","_number":1,"created":"2018-07-06 18:53:29.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/41/580741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/41/580741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/41/580741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/41/580741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/41/580741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/41/580741/1"}}},"commit":{"parents":[{"commit":"a6221f558ac9bfc1e0fc326c09711bb7891ac97f","subject":"switch to PTI documentation jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/a6221f558ac9bfc1e0fc326c09711bb7891ac97f"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-07-06 18:52:27.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-07-06 18:52:27.000000000","tz":-240},"subject":"fix bug link in readme","message":"fix bug link in readme\n\nThis patch fixes the bug tracker link in the readme. It is set to\ndepend on a job definition change in project-config so we can use this\npatch to test the new release jobs.\n\nChange-Id: I1e0aeae7b0cbc44f4b5d2054fec91ea5552fda22\nDepends-On: https://review.openstack.org/580740\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/288487af7c7c58c439efb22fec5ba6d1fbc47feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/288487af7c7c58c439efb22fec5ba6d1fbc47feb"}]},"branch":"refs/heads/master"},"e9788f4f48c09d9e5869e4279110c88af58c4216":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-07-09 17:51:06.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/41/580741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/41/580741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/41/580741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/41/580741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/41/580741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/41/580741/2"}}},"commit":{"parents":[{"commit":"b1292a23f190461cf6e60961aa3adb23a7a973d5","subject":"switch to PTI documentation jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/b1292a23f190461cf6e60961aa3adb23a7a973d5"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-07-06 18:52:27.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-07-09 17:51:00.000000000","tz":-240},"subject":"fix bug link in readme","message":"fix bug link in readme\n\nThis patch fixes the bug tracker link in the readme. It is set to\ndepend on a job definition change in project-config so we can use this\npatch to test the new release jobs.\n\nChange-Id: I1e0aeae7b0cbc44f4b5d2054fec91ea5552fda22\nDepends-On: https://review.openstack.org/580740\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/e9788f4f48c09d9e5869e4279110c88af58c4216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/e9788f4f48c09d9e5869e4279110c88af58c4216"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
