)]}'
{"id":"openstack%2Fwhereto~667084","triplet_id":"openstack%2Fwhereto~master~I74ae3e38f62cb4fd4dc57af39480c6c95e426878","project":"openstack/whereto","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I74ae3e38f62cb4fd4dc57af39480c6c95e426878","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-06-24 13:15:00.000000000","updated":"2019-06-24 14:53:18.000000000","submitted":"2019-06-24 14:53:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667084-1561387998648-6b32c878","meta_rev_id":"d7c84fa5bdb8e7f3172bc5e99c8d80894ceed066","_number":667084,"virtual_id_number":667084,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-06-24 14:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-06-24 14:23:52.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-06-24 14:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-06-24 14:44:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 14:23:52.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-06-24 14:44:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-06-24 14:53:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4c5e3d622bd217494e3c43211062e78237a9504","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-06-24 13:15:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a18573243502d89d3dd6ac417a1e4b9021494c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 13:25:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/84/667084/1/check/test-release-openstack/475b7d4/ : SUCCESS in 3m 11s\n- openstack-tox-pep8 http://logs.openstack.org/84/667084/1/check/openstack-tox-pep8/5d762c9/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/84/667084/1/check/openstack-tox-py27/641b136/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/84/667084/1/check/openstack-tox-py36/fe15767/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/84/667084/1/check/openstack-tox-py37/b4566d5/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/84/667084/1/check/openstack-tox-docs/b08b61c/html/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c4c6190bb6550883acf0d89f48c2c51d0d2ea64d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-24 14:23:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6527640eb160724563db1e733c07306d599ede3f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-24 14:44:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35a910109300d073aa6af3e50850e9e945c8b5ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:44:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59f38e8d409ccdfd89e10c181a3bf4f75f6f8dfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:53:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b3b1e440c5aa76e3ddb5ff713aacd5dea1f6267c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:53:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/84/667084/1/gate/test-release-openstack/b06f6a6/ : SUCCESS in 2m 25s\n- openstack-tox-pep8 http://logs.openstack.org/84/667084/1/gate/openstack-tox-pep8/8582a18/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/84/667084/1/gate/openstack-tox-py27/ae85f5f/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/84/667084/1/gate/openstack-tox-py36/01d495b/ : SUCCESS in 3m 05s\n- openstack-tox-py37 http://logs.openstack.org/84/667084/1/gate/openstack-tox-py37/9892459/ : SUCCESS in 3m 07s\n- openstack-tox-docs http://logs.openstack.org/84/667084/1/gate/openstack-tox-docs/15f870a/html/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa1c5eb1a10750b9c6ac447f5eeedd542461899f","revisions":{"fa1c5eb1a10750b9c6ac447f5eeedd542461899f":{"kind":"REWORK","_number":1,"created":"2019-06-24 13:15:00.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/84/667084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/84/667084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/84/667084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/84/667084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/84/667084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/84/667084/1"}}},"commit":{"parents":[{"commit":"d61b279ab01a0735694f20f6e6637a00c5b94563","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/d61b279ab01a0735694f20f6e6637a00c5b94563"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-24 13:12:27.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-24 13:14:59.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I74ae3e38f62cb4fd4dc57af39480c6c95e426878\nStory: #2005924\nTask: #34205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/fa1c5eb1a10750b9c6ac447f5eeedd542461899f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/fa1c5eb1a10750b9c6ac447f5eeedd542461899f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
