)]}'
{"id":"openstack%2Fwhereto~674198","triplet_id":"openstack%2Fwhereto~master~I6af9bdb5968c00e23e272b983839380f2f9f17cb","project":"openstack/whereto","branch":"master","topic":"sphinx","hashtags":[],"change_id":"I6af9bdb5968c00e23e272b983839380f2f9f17cb","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","status":"MERGED","created":"2019-08-02 06:35:45.000000000","updated":"2019-08-04 13:41:33.000000000","submitted":"2019-08-04 13:41:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674198-1564926093466-efd57ca1","meta_rev_id":"e0b3d544fad03926ea2537c52429def7e571add8","_number":674198,"virtual_id_number":674198,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-04 13:41:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-08-02 07:47:59.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2019-08-04 13:32:56.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-04 13:32:56.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-04 13:41:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97372498c0ea5b122b8234283812d7c97c5efbb2","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-02 06:35:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f0bdf5ad4be8ebe968ebca67aa39e341df79787","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 06:46:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/98/674198/1/check/openstack-tox-pep8/855e89e/ : SUCCESS in 5m 19s\n- openstack-tox-py27 https://logs.opendev.org/98/674198/1/check/openstack-tox-py27/17d8b07/ : SUCCESS in 5m 45s\n- openstack-tox-py36 https://logs.opendev.org/98/674198/1/check/openstack-tox-py36/9e9b6ca/ : SUCCESS in 4m 39s\n- openstack-tox-py37 https://logs.opendev.org/98/674198/1/check/openstack-tox-py37/bd8e357/ : SUCCESS in 3m 36s\n- openstack-tox-docs https://logs.opendev.org/98/674198/1/check/openstack-tox-docs/15c740a/html/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"53d402971f3073f08898c0ccd50ec75365b97b05","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-02 07:47:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0feaa5ff142a8571083f32f0ba1b3567fdbbdf28","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-04 13:32:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"59bfd27a54d875dc43b6eeac245d61a73b8d23ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 13:33:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1ef5960589f15628ca79450bc58eef599721572","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 13:41:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/98/674198/1/gate/openstack-tox-pep8/06d0801/ : SUCCESS in 2m 24s\n- openstack-tox-py27 https://logs.opendev.org/98/674198/1/gate/openstack-tox-py27/95d30a7/ : SUCCESS in 2m 43s\n- openstack-tox-py36 https://logs.opendev.org/98/674198/1/gate/openstack-tox-py36/222bf61/ : SUCCESS in 2m 58s\n- openstack-tox-py37 https://logs.opendev.org/98/674198/1/gate/openstack-tox-py37/fc56159/ : SUCCESS in 3m 26s\n- openstack-tox-docs https://logs.opendev.org/98/674198/1/gate/openstack-tox-docs/86636a3/html/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"cd0d9dc99daa2505f9c0da21cc10f67d7948361d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 13:41:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fce8065ac95cc97fe82f0d65cf27b465ae5f3190","revisions":{"fce8065ac95cc97fe82f0d65cf27b465ae5f3190":{"kind":"REWORK","_number":1,"created":"2019-08-02 06:35:45.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/98/674198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/98/674198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/98/674198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/98/674198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/98/674198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/98/674198/1"}}},"commit":{"parents":[{"commit":"ce0ea965cb062e0995688c6898cc51642d293e60","subject":"Remove modindex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/ce0ea965cb062e0995688c6898cc51642d293e60"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 06:35:18.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 06:36:01.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\n\nChange-Id: I6af9bdb5968c00e23e272b983839380f2f9f17cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/fce8065ac95cc97fe82f0d65cf27b465ae5f3190"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/fce8065ac95cc97fe82f0d65cf27b465ae5f3190"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
