)]}'
{"id":"openstack%2Fwhereto~744338","triplet_id":"openstack%2Fwhereto~master~I8c862d12a9e5f5fb588db667845b94f81fbabbcb","project":"openstack/whereto","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I8c862d12a9e5f5fb588db667845b94f81fbabbcb","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-01 23:08:34.000000000","updated":"2020-08-02 23:28:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"609de31ae899b21568200486e149ff8b6612f4bf","_number":744338,"virtual_id_number":744338,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-02 21:22:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-02 21:22:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"edb1009cb09fc6c3e4967780c9230b2ac0c3dcaa","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-01 23:08:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f26b143e1a024b8c76e760409c00332069d21af2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 23:18:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d71cb5647fdd46a1a1c0dc46a8fcbf90 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5534e50640cf4f5796df977452db67e9 : SUCCESS in 2m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ab8372b224e452dab4fc0119f01a647 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3455afbc1174a74b8e1ab73826f614c : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb665473b2f54368a66317ea31c24562 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"405c1c21124275c7c01bcfecc120c37ef6fd40b8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 19:47:03.000000000","message":"Abandoned\n\nall working here, no updates needed for Focal miration.","accounts_in_message":[],"_revision_number":1},{"id":"67ee6d936bf93a8a1c4cdaade79153fef3a1a614","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 20:04:52.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"226be871b51413c94b45983ec3bc604c90895f0c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 20:05:07.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"adb8a5a251299d21642d5f8cfd05098021425c91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 21:22:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2ec04a531da440b88a599a57febe652 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67ad2deab4044754a6117e0a925169c1 : SUCCESS in 2m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3abadc48f1eb449884e6744ab9dc75b7 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fc4b903798e44acb0794350a91be64a : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2ae1d424be4158b188e9974d3ce487 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"19e6d1464b9a4dd7b7854e56aad93e8bf4cfb4a0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 23:28:35.000000000","message":"Abandoned\n\nall working here, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"30e29cb498e07c3dec8fdf50fcfe15f6c834c425","revisions":{"342efb1cf21d2185ace517e5abeea40c6d578c70":{"kind":"REWORK","_number":1,"created":"2020-08-01 23:08:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/38/744338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/38/744338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/744338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/744338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/744338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/38/744338/1"}}},"commit":{"parents":[{"commit":"cd00454b5362095e03a8173e15108bd63923f863","subject":"Merge \"Switch to newer openstackdocstheme version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/cd00454b5362095e03a8173e15108bd63923f863"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 23:08:15.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 23:08:32.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nChange-Id: I8c862d12a9e5f5fb588db667845b94f81fbabbcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/342efb1cf21d2185ace517e5abeea40c6d578c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/342efb1cf21d2185ace517e5abeea40c6d578c70"}]},"branch":"refs/heads/master"},"30e29cb498e07c3dec8fdf50fcfe15f6c834c425":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-02 20:05:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/38/744338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/38/744338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/744338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/744338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/38/744338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/38/744338/2"}}},"commit":{"parents":[{"commit":"cd00454b5362095e03a8173e15108bd63923f863","subject":"Merge \"Switch to newer openstackdocstheme version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/cd00454b5362095e03a8173e15108bd63923f863"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 23:08:15.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 20:05:04.000000000","tz":0},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nChange-Id: I8c862d12a9e5f5fb588db667845b94f81fbabbcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/30e29cb498e07c3dec8fdf50fcfe15f6c834c425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/30e29cb498e07c3dec8fdf50fcfe15f6c834c425"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
