)]}'
{"id":"openstack%2Fwhereto~788171","triplet_id":"openstack%2Fwhereto~master~I35e5dca33aeff98e31ddd25baecb58ccf4078890","project":"openstack/whereto","branch":"master","hashtags":[],"change_id":"I35e5dca33aeff98e31ddd25baecb58ccf4078890","subject":"setup.cfg: Replace dashes with underscores","status":"ABANDONED","created":"2021-04-27 06:53:27.000000000","updated":"2022-08-04 16:35:38.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"55b6d6a549a66f29c40d635167809a4f1f6dfce3","_number":788171,"virtual_id_number":788171,"owner":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-27 07:04:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 07:04:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-27 14:18:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"}],"messages":[{"id":"efd0fe333850e4d414975c02a58ab0b0892fa2ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2021-04-27 06:53:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af04975483e16573c32bbc77c60598f2466c17ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 07:04:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/557a8f249cb948fa96f448208802daf3 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88b2c3322d504a72816ae49a2f03eff7 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d3773776d2944ccafbaa07dac9f4568 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a00f03278994c08936657d81e0c1815 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe3cfbf9cc444f21b931f1054a570c82 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41904efcbf0440ad9558c56fdf250838 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8698b9efae4fadf75c4e9110df43c548a1fae97d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-27 14:18:16.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nHello,\n\nI wonder if these options aren\u0027t proxified by pbr [1], and so, if we need that.\n\nHowever, these changes can\u0027t hurt even if pbr is behind our packaging, I think it won\u0027t be an issue for pbr. \n\nAlso please have a look to my inline comment.\n\n[1] https://opendev.org/openstack/pbr/src/branch/master/pbr/util.py#L95-L137","accounts_in_message":[],"_revision_number":1},{"id":"55b6d6a549a66f29c40d635167809a4f1f6dfce3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-04 16:35:38.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/whereto/+/852000 is being merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1414fb13557e8ac8d3684e606f2a17b7008e6c8b","revisions":{"1414fb13557e8ac8d3684e606f2a17b7008e6c8b":{"kind":"REWORK","_number":1,"created":"2021-04-27 06:53:27.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/71/788171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/71/788171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/71/788171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/71/788171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/71/788171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/71/788171/1"}}},"commit":{"parents":[{"commit":"492c33675f62dc34703c7486b15a2da2d062de9c","subject":"Migrate from testr to stestr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/492c33675f62dc34703c7486b15a2da2d062de9c"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2021-04-27 06:52:51.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2021-04-27 06:53:22.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I35e5dca33aeff98e31ddd25baecb58ccf4078890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/1414fb13557e8ac8d3684e606f2a17b7008e6c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/1414fb13557e8ac8d3684e606f2a17b7008e6c8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
