)]}'
{"id":"openstack%2Fwhereto~825326","triplet_id":"openstack%2Fwhereto~master~I3da0593718fa0196e238a2d5a343b958ce34f201","project":"openstack/whereto","branch":"master","attention_set":{},"removed_from_attention_set":{"33881":{"account":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"last_update":"2022-03-22 12:22:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3da0593718fa0196e238a2d5a343b958ce34f201","subject":"remove unicode literal from code","status":"MERGED","created":"2022-01-19 15:01:18.000000000","updated":"2022-03-22 12:24:14.000000000","submitted":"2022-03-22 12:22:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825326","meta_rev_id":"e06d5eb5d37303926997b9ede07c76282029908c","_number":825326,"virtual_id_number":825326,"owner":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-22 12:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-02-24 10:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-03-22 12:09:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 15:15:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 10:30:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"d685af815002ffee7ff445afcecb0a1b47b57a7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-01-19 15:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b455968a66a065f21bedd4cb4eb8eb6411e6e2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 15:15:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9a9df97d4564bea95d5a4578103a7a6 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd04aa12257e45bd9c015a3c31495708 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fbba82f490a43fea0a106d198f6a9e6 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d9cd6c2c68349d3a01dfd5577bc662e : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dace02bab564dce8030176949a8c87b : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3b04a5430f5d3734a1c9ae096f18ae3b99522d4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-02-24 00:54:06.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"39c4af2579185d0fc0061529cdc042e2dcc87725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 00:59:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34205a93bb834470931d7f141ae03e46 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4415f2e2575445bc9e021b4d02c27a3d : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af130e5c651d48229fd5107a3c0389b7 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22342e99d5ad482197db453fbd269d0e : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae857c069af84b868ebfa148061c32dd : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"814b443252ce2c864899e3093821c537a3f3bc82","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-24 10:30:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"74154d534a138ee2b08dc6b12356f930743b0dac","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-22 12:09:38.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"79c94f3309b2610161e716dc6d7c031617aba72f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:10:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5306f68b15af4669a4cecf19de64908c317f908e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:22:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/347ab3cdbd634d3cb8f5005eb9980dae : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca026f35e85a42de8019aa74bfd3cd02 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea732baef41b4480ae06f2d47ff77218 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16930f56d8374096aae5c0f036156b8c : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"1f0b0e35fd3dabdfdf2ec48d4785f57d96dd154a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:22:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e06d5eb5d37303926997b9ede07c76282029908c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:24:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/536111ce70c34874bfceb8f84245791c : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81ec6a930f125a1086da31b26c2b81753b3c7458","revisions":{"d75f573d8be414002422e6890cd45ea631e17c1f":{"kind":"REWORK","_number":1,"created":"2022-01-19 15:01:18.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/26/825326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/26/825326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/26/825326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/26/825326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/26/825326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/26/825326/1"}}},"commit":{"parents":[{"commit":"492c33675f62dc34703c7486b15a2da2d062de9c","subject":"Migrate from testr to stestr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/492c33675f62dc34703c7486b15a2da2d062de9c"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-01-19 15:00:33.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-01-19 15:00:33.000000000","tz":480},"subject":"Remove unicode from code","message":"Remove unicode from code\n\nChange-Id: I3da0593718fa0196e238a2d5a343b958ce34f201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/d75f573d8be414002422e6890cd45ea631e17c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/d75f573d8be414002422e6890cd45ea631e17c1f"}]},"branch":"refs/heads/master"},"81ec6a930f125a1086da31b26c2b81753b3c7458":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-24 00:54:06.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/26/825326/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/26/825326/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/26/825326/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/26/825326/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/26/825326/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/26/825326/2"}}},"commit":{"parents":[{"commit":"492c33675f62dc34703c7486b15a2da2d062de9c","subject":"Migrate from testr to stestr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/492c33675f62dc34703c7486b15a2da2d062de9c"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-01-19 15:00:33.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-02-24 00:54:06.000000000","tz":0},"subject":"remove unicode literal from code","message":"remove unicode literal from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I3da0593718fa0196e238a2d5a343b958ce34f201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/81ec6a930f125a1086da31b26c2b81753b3c7458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/81ec6a930f125a1086da31b26c2b81753b3c7458"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
