)]}'
{"id":"openstack%2Fwhereto~825329","triplet_id":"openstack%2Fwhereto~master~Id6b12adb05695fc4ac118e1529a1bf58219c19a6","project":"openstack/whereto","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-08-04 16:36:43.000000000","reason":"Change was abandoned"},"33881":{"account":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"last_update":"2022-08-04 16:36:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id6b12adb05695fc4ac118e1529a1bf58219c19a6","subject":"Update python testing classifier","status":"ABANDONED","created":"2022-01-19 15:16:54.000000000","updated":"2022-08-04 16:36:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2b0cccea48d96375e246f64e427cbc02e0fc08fa","_number":825329,"virtual_id_number":825329,"owner":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2022-03-22 12:29:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-03-22 12:18:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-03-22 12:18:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 15:30:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-22 12:18:42.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"ea5916eca7cf987bb026f0f384a2d91147c1e9e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-01-19 15:16:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7029bea22d0499c3cb4329450a0b55626165e26d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 15:30:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d0818d43e32a4ae580351d8704696abd : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc8d27d3c81849fc85f7ad46571c3bea : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44593460d0df415d95775dad87c01ad8 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/544f607f94f14e37b5256c9ee1756fa0 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/285c845aa3fa4161a52d5a396e4252e1 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad20b34db644135a3eb529e2039831b : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"fad278fcd73380412f27aaade91dafa8da230fe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-22 12:18:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3d3c207f3d5c90d43e618265ade9efbb80547b93","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-22 12:18:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"68d87e25682e03a801e6ea9884dcf271bb128810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:28:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bbcefe7f859a478ca2400ac50f8bd837 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a3aea879094100ae5c4feaf57d0ebf : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cfc1f6f088548e6a02dadf535021d27 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c30ad348025414c9614cd3e12edc2dc : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb3ae61823dc40199068174753d612e8 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a3688e51664821aa77be7a36201609 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"2f6213743f57e0ed77f4ac2ad9f9465d147db52b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:29:01.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/whereto for 825329,2","accounts_in_message":[],"_revision_number":2},{"id":"2b0cccea48d96375e246f64e427cbc02e0fc08fa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-04 16:36:43.000000000","message":"Abandoned\n\nwe no longer need this, it seems","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f82bfd887c233751a747b254c145aff574b2740f","revisions":{"b1eb12c64fda637ab5e27fc46ce15329dc33aa59":{"kind":"REWORK","_number":1,"created":"2022-01-19 15:16:54.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/29/825329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/29/825329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/29/825329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/29/825329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/29/825329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/29/825329/1"}}},"commit":{"parents":[{"commit":"d75f573d8be414002422e6890cd45ea631e17c1f","subject":"Remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/d75f573d8be414002422e6890cd45ea631e17c1f"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-01-19 15:11:59.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-01-19 15:11:59.000000000","tz":480},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nChange-Id: Id6b12adb05695fc4ac118e1529a1bf58219c19a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/b1eb12c64fda637ab5e27fc46ce15329dc33aa59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/b1eb12c64fda637ab5e27fc46ce15329dc33aa59"}]},"branch":"refs/heads/master"},"f82bfd887c233751a747b254c145aff574b2740f":{"kind":"NO_CHANGE","_number":2,"created":"2022-03-22 12:18:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/29/825329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/29/825329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/29/825329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/29/825329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/29/825329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/29/825329/2"}}},"commit":{"parents":[{"commit":"81ec6a930f125a1086da31b26c2b81753b3c7458","subject":"remove unicode literal from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/81ec6a930f125a1086da31b26c2b81753b3c7458"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-01-19 15:11:59.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-03-22 12:18:36.000000000","tz":0},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nChange-Id: Id6b12adb05695fc4ac118e1529a1bf58219c19a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/f82bfd887c233751a747b254c145aff574b2740f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/f82bfd887c233751a747b254c145aff574b2740f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
