)]}'
{"id":"openstack%2Fwhereto~828468","triplet_id":"openstack%2Fwhereto~master~I84d8f96178486ec694b7732dbce637b302c7278f","project":"openstack/whereto","branch":"master","attention_set":{},"removed_from_attention_set":{"33881":{"account":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"last_update":"2022-03-22 12:49:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I84d8f96178486ec694b7732dbce637b302c7278f","subject":"Use TOX_CONSTRAINTS_FILE","status":"MERGED","created":"2022-02-09 07:54:33.000000000","updated":"2022-03-22 12:51:01.000000000","submitted":"2022-03-22 12:49:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828468","meta_rev_id":"b77f1844992ed22c731bfc937b98e6daedeeb073","_number":828468,"virtual_id_number":828468,"owner":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-22 12:49:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-03-22 12:18:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-22 12:51:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-03-22 12:18:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 08:05:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 13:16:06.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2022-03-22 12:14:01.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"c7948db03a51be09ed1e75ffe38351987044dbf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-02-09 07:54:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0d9bb8837822cd1e59ac4fb17333d302400fe4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-02-09 07:55:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"223b9fd1889b34a30bee6b3588074c597ad55491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 08:05:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2ee3ce27b0349199788787032360d02 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d298b2e60b1b46c5a2604473ae880f80 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c05d71583e904438bd915fb1b52be177 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15f5ddad6a88454ea47120c8a644cf94 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c304dcdf3bb04487a16169a01ed14d6c : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"266e41a28692f6d2775f8673dda76a4ae15276a8","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-02-10 13:16:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4b33d0cbd6e24b4574ec4eb6b8c334d1f60cb66c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-22 12:14:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"03d06b4507c9f4bd4a371941262a6ea55c4a60b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-22 12:18:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d4216a282293bba5797461ee0f4709c06dd6bbf9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-22 12:18:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3aa73730d120bd11b77794a29e440aa5b6476417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:38:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ae734b39f744164abf1a5abf6b79268 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b384d0b4d6344488b52f6ca49138d87a : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c10b376f2a44f588d90b6b5235cf1a9 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acd24a1485a94ffda40f00a03acd8d53 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4efb1345c14411fac8c8f672ad9d860 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":3},{"id":"e1c98afbde9dafc71a05a7bcdf93161642c52171","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:39:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1d8890073137b350bf2998b912aaeba7edb59f2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:49:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c36052cacf4447c08758e68873b2a50d : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b06c215848440cdbf932305ff3f11c2 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98f156a212bd4d7baf71838939bb531d : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fea2b823e44d8f96b5ed854576ea53 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":3},{"id":"ec4e14e6475a9eb436fb73d234754c7983dc470b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:49:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8f90983d3e4500feda4b9fba5c291e9b7b34344c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:51:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec50afc84d5422a8cbfb8dea8229506 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9f28d5654a4adc50ccbba217344f7d0381f220fd","revisions":{"e322f7629d4e8409bfb13bd384a1141abda3cc79":{"kind":"REWORK","_number":1,"created":"2022-02-09 07:54:33.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/68/828468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/68/828468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/68/828468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/68/828468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/68/828468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/68/828468/1"}}},"commit":{"parents":[{"commit":"b1eb12c64fda637ab5e27fc46ce15329dc33aa59","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/b1eb12c64fda637ab5e27fc46ce15329dc33aa59"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-02-09 07:54:02.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-02-09 07:54:02.000000000","tz":480},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nChange-Id: I84d8f96178486ec694b7732dbce637b302c7278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/e322f7629d4e8409bfb13bd384a1141abda3cc79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/e322f7629d4e8409bfb13bd384a1141abda3cc79"}]},"branch":"refs/heads/master"},"a44f550449c7800d52e8dc5fdd6f5929be4bda3a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-09 07:55:57.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/68/828468/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/68/828468/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/68/828468/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/68/828468/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/68/828468/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/68/828468/2"}}},"commit":{"parents":[{"commit":"b1eb12c64fda637ab5e27fc46ce15329dc33aa59","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/b1eb12c64fda637ab5e27fc46ce15329dc33aa59"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-02-09 07:54:02.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-02-09 07:55:57.000000000","tz":0},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I84d8f96178486ec694b7732dbce637b302c7278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/a44f550449c7800d52e8dc5fdd6f5929be4bda3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/a44f550449c7800d52e8dc5fdd6f5929be4bda3a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9f28d5654a4adc50ccbba217344f7d0381f220fd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-22 12:18:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/828468/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/68/828468/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/68/828468/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/68/828468/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/68/828468/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/68/828468/3"}}},"commit":{"parents":[{"commit":"ba96f9f3307c36e62ccc5d98034456c039f4e937","subject":"Update CI to use unversioned jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/ba96f9f3307c36e62ccc5d98034456c039f4e937"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-02-09 07:54:02.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-03-22 12:18:52.000000000","tz":0},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I84d8f96178486ec694b7732dbce637b302c7278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/9f28d5654a4adc50ccbba217344f7d0381f220fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/9f28d5654a4adc50ccbba217344f7d0381f220fd"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
