)]}'
{"id":"openstack%2Fwhereto~935408","triplet_id":"openstack%2Fwhereto~master~Ica7215d747f9a19211f5ca6ad8eb107cca00c7ee","project":"openstack/whereto","branch":"master","topic":"migrate-to-noble","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-02-05 17:57:09.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Ica7215d747f9a19211f5ca6ad8eb107cca00c7ee","subject":"DNM: Testing ubuntu noble","status":"ABANDONED","created":"2024-11-15 22:10:14.000000000","updated":"2025-03-11 10:59:10.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7bbe8e676a73505a93d1897a6869292f5963ea01","_number":935408,"virtual_id_number":935408,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-05 18:04:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-15 22:24:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"692a15efe2dcb36bc436cd108f8197e805248b50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-15 22:10:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"236905eb5493b5b080a95ecf504848fbe025f8fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 22:24:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32913d85cabc4115856a4f9ed7425611\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/344a42e568e443cd903c4fcd9b4dc03a : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec1f18a0753d4b7ba722dbda104c71d0 : SUCCESS in 2m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b80d7c920b5f461cb62fcca8166e2312 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75afe1866dc64a86a36f1317577fae95 : FAILURE in 2m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c4beed4cf36e414fa59a5e1fa87cf0fa6ac78ee5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-16 04:28:08.000000000","message":"Topic set to migrate-to-noble","accounts_in_message":[],"_revision_number":1},{"id":"a9e1a0b8ed7ef329ca173b8f829ccf8bb87bc723","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-17 04:39:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6553a177fe97f38430a4deb30cfc91e0a28939df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-17 04:45:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a93df3e649db45ad909c13b15d87bdfa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/340d86769bdf456589d6d5e2b3114dfc : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c893c523097246978af04682a8e8a8dd : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bc18e8aa9124d8b8efa860e8a33b4ad : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e456858503f4171b0ee6152c448a391 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7a7f8cf2cd0f16ada6d355abf7e5c786a9ca1cd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-20 19:46:23.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d17412d5a48fd1283af6ea39f534714b8012fa7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 19:56:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f5587ebc9c34ca49def0fdb12bf99dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94336fb61ace4f7989b3a7bbbe61883f : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e89d878cbe64b6680243d0cb5b779d8 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3267a4c5aec84227913f968fc22b0fa9 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe85a07a07a94212a64a79b15cff60ab : FAILURE in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1d9c277a85811157a0833e7e421e1140c5c09c76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-24 18:37:05.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b968cbba3f180b69ee8fcf0339673e4e3240357b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 18:46:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f74139e4df424a86882acd39811924e3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012df505d58a4f50bf824bb737bd51e5 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/367a870f0a654a509a87bd982b32f726 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2566eb01182c4a7989433e90710683d4 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8561d5ea45684754b3b985212576dd55 : FAILURE in 2m 03s","accounts_in_message":[],"_revision_number":3},{"id":"2137c30c649e8f1b9d185d1450a983a5cb8a5b5d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-27 18:05:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3d26f2357b81ea006d8a8125838d15633c9c56c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 18:14:03.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc8499660d0f4fb48ee8c4822bfcf4a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf8489fa5514737b3741332a16f079b : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f195e11c64c842a0a90e7a1f6854e473 : SUCCESS in 2m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae39179b5ad24431aca8fbb725b8026e : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eaa16565570424b8e8e5413f783d264 : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":3},{"id":"008a0bc333cf40faf78c4431416b5fab29f05780","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-28 19:58:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82c20ccf2ba2f687561633a17904717ce5571502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 20:09:05.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02841bce360842439db8d23ad965cc80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50abde3b51cb4efd892a21f20ca4dde8 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3226e1edc2fb40de8cc73b29cf622f5c : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bc77a9992554aee83bd3c92f04c9355 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22dca1c8a9474238bfa2bbbaf3ce23ca : FAILURE in 3m 30s","accounts_in_message":[],"_revision_number":3},{"id":"b9d784b2e7b18ee8490c0f5803850b45884353be","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-05 17:57:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d6bb71d821ce85de31b00f1d43a96daf96764f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 18:04:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2026dc970784edaaf0e50aca22f526c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd15b1ebbaee43d0979fb56969787ec2 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c11dd89de02c45f39ce8a0989974a3e6 : SUCCESS in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2bfa9dac60b94a13b2090dcf99629779 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14147bd486654095ab5ac8e298c83fbe : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":3},{"id":"0a14491002007f7ac6b6d651b913ceacfce0a651","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-05 18:33:20.000000000","message":"Abandoned\n\nall green here,","accounts_in_message":[],"_revision_number":3},{"id":"c3adacbe7e0a0431a40d72f0beb8dee5923d1164","tag":"autogenerated:gerrit:restore","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-03-11 05:33:22.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"76fc5076d7b9259f3c55c35fd082f7a68c05cb91","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-03-11 05:33:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b820bde37816aa36b10b71cc0f6a719c106c2c0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 06:05:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6177627b593d45578a486b19512e4b4d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39403c8429a44beaa1d92a1d2cef6e47 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d789857993164c66be790d8a4b6d8f4b : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f2076e1664643b586ea8077b538a29b : SUCCESS in 1m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b90ce5edf64a1798acf7ca22a954a0 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":3},{"id":"7bbe8e676a73505a93d1897a6869292f5963ea01","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-03-11 10:59:10.000000000","message":"Abandoned\n\nall good, doc job is passing on Noble","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7bbc6f0d388cd009d1faaa34dff0fb4d254d15df","revisions":{"0e5b65b0632f236d9bf844547da71294495c0655":{"kind":"REWORK","_number":1,"created":"2024-11-15 22:10:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/08/935408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/08/935408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/08/935408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/08/935408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/08/935408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/08/935408/1"}}},"commit":{"parents":[{"commit":"7185959eb00692ca34b2229760b3502cc9116874","subject":"Drop unnecessary \u0027x\u0027 bit from doc config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/7185959eb00692ca34b2229760b3502cc9116874"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-15 22:08:06.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-15 22:10:12.000000000","tz":-480},"subject":"DNM: Testing ubuntu noble","message":"DNM: Testing ubuntu noble\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/932155\nChange-Id: Ica7215d747f9a19211f5ca6ad8eb107cca00c7ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/0e5b65b0632f236d9bf844547da71294495c0655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/0e5b65b0632f236d9bf844547da71294495c0655"}]},"branch":"refs/heads/master"},"d53b62cbaeab21df98cd2dbe1155584708d53153":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-12-20 19:46:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/08/935408/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/08/935408/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/08/935408/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/08/935408/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/08/935408/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/08/935408/2"}}},"commit":{"parents":[{"commit":"7185959eb00692ca34b2229760b3502cc9116874","subject":"Drop unnecessary \u0027x\u0027 bit from doc config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/7185959eb00692ca34b2229760b3502cc9116874"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-15 22:08:06.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-12-20 19:46:23.000000000","tz":0},"subject":"DNM: Testing ubuntu noble","message":"DNM: Testing ubuntu noble\n\nDepends-On: https://review.opendev.org/c/openstack/openstackdocstheme/+/935460\nChange-Id: Ica7215d747f9a19211f5ca6ad8eb107cca00c7ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/d53b62cbaeab21df98cd2dbe1155584708d53153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/d53b62cbaeab21df98cd2dbe1155584708d53153"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7bbc6f0d388cd009d1faaa34dff0fb4d254d15df":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-01-24 18:37:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/08/935408/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/08/935408/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/08/935408/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/08/935408/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/08/935408/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/08/935408/3"}}},"commit":{"parents":[{"commit":"7185959eb00692ca34b2229760b3502cc9116874","subject":"Drop unnecessary \u0027x\u0027 bit from doc config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/7185959eb00692ca34b2229760b3502cc9116874"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-15 22:08:06.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2025-01-24 18:37:05.000000000","tz":0},"subject":"DNM: Testing ubuntu noble","message":"DNM: Testing ubuntu noble\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/935459/\nChange-Id: Ica7215d747f9a19211f5ca6ad8eb107cca00c7ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/7bbc6f0d388cd009d1faaa34dff0fb4d254d15df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/7bbc6f0d388cd009d1faaa34dff0fb4d254d15df"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
