)]}'
{"id":"openstack%2Fwhereto~970061","triplet_id":"openstack%2Fwhereto~master~I7725cdb8af787bbaa387f1235d128f28eab3719f","project":"openstack/whereto","branch":"master","topic":"pcre","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-11 13:01:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7725cdb8af787bbaa387f1235d128f28eab3719f","subject":"Replace obsolete PCRE packages","status":"MERGED","created":"2025-12-07 16:10:52.000000000","updated":"2025-12-11 13:03:27.000000000","submitted":"2025-12-11 13:01:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970061-pcre","meta_rev_id":"290a45e15c414140888491defd62dc35f22db9f9","_number":970061,"virtual_id_number":970061,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-11 13:01:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-12-11 12:51:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-12-08 09:08:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-12-11 12:51:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-07 16:27:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-08 09:08:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-12-11 12:51:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"260eb90f68e851f6bec7d38fc6e2d82e9df6fb81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 16:10:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a29f20c93f268f7d94b10b7e8e633afaae647f9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 16:18:23.000000000","message":"Topic set to pcre","accounts_in_message":[],"_revision_number":1},{"id":"8e75f8417155704762fe05e68b20fe8c25995d06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 16:18:34.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"67849271a13bc8f521639b1d519b64dc209c1969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 16:27:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d4a95a0e67242539483395e80a18e21\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/298a844dd84b458894eac112915532c2 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e092aac57a94486924a04177b06e096 : SUCCESS in 1m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59a6a5ced22c457aa679cbc3f6e4ae93 : FAILURE in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9f4728165544a5fb7008260f5bd5e0d : FAILURE in 2m 24s","accounts_in_message":[],"_revision_number":2},{"id":"9b626ba2ff574dc785cfa780692c7fb8ed85360d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 16:29:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9e9fa380c8e140686e2890bc645530ce04d42baf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 16:36:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b38b895884b44721a508de7e1a3da9ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f33644b06544c31baa7f3b092bdd569 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c444c20f7e094adba2a2c8d56a706f09 : FAILURE in 1m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f9581ee5c0346f898895c3fb5897cc9 : FAILURE in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf80c1f73174011be795265dae10bbd : FAILURE in 1m 31s","accounts_in_message":[],"_revision_number":3},{"id":"ba9387899d2b8f08492d5e46d78a48b4f7d2329f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 17:18:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9470a228e985b7c1291092f9684ba166ce09f8b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 17:23:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a8a13bbc98c4fde8a8e41c7d751df05\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85a0486655134a3098b3188833368efa : SUCCESS in 2m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ba50cf02e4e41e89e876ee809bd1ea1 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f25ad1a9d37b4183accaa8273e9da3c0 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a4529d29f547adb7900cc0713288d4 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":4},{"id":"e69c888fdba66fc5c5d84f7c3975aeb737153f5b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-12-08 09:08:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d128a43b71c3fa54b99337513c19e0faeab4c365","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-11 12:51:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"21d7276d1404605401c8df6b7a5f777c9eee52f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 12:52:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b77021389b5d23c131cc83cc1a47fd7f82679608","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 13:01:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/365e658134934c28af8694f9c1952d15\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7d2b21471f94f1c973f0cf7ac033d3a : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ef8598c65614ec5bca1ed515f59f96f : SUCCESS in 2m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52a78a30571c4a07917a81de8d176077 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aea9185d194a4b0180f3f8e31fc271ef : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":4},{"id":"603d222e2cb19ecf41aa14536845b4b6d7c9e00f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 13:01:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"290a45e15c414140888491defd62dc35f22db9f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 13:03:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c33aeed0f9240b389ebd903aec10257\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a82ed4a21894113b62313d5651e3e33 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"061b8cedde04a41d28f431d812ee0c5fc329ee14","revisions":{"cebd842123bb8ed1792b0e1b7af51cbd63633869":{"kind":"REWORK","_number":1,"created":"2025-12-07 16:10:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/970061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/61/970061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/61/970061/1"}}},"commit":{"parents":[{"commit":"77829ccdb474c4755be83ef6bdf09430906b87ab","subject":"Merge \"Drop basepython\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/77829ccdb474c4755be83ef6bdf09430906b87ab"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 16:10:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 16:10:43.000000000","tz":540},"subject":"Replace obsolete PCRE3 package","message":"Replace obsolete PCRE3 package\n\nIt was replaced by PCRE2 library.\n\nChange-Id: I7725cdb8af787bbaa387f1235d128f28eab3719f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/cebd842123bb8ed1792b0e1b7af51cbd63633869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/cebd842123bb8ed1792b0e1b7af51cbd63633869"}]},"branch":"refs/heads/master"},"13a6285ec12dc343211f0ddfe418c951e39dcd84":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-07 16:18:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/970061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/61/970061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/61/970061/2"}}},"commit":{"parents":[{"commit":"77829ccdb474c4755be83ef6bdf09430906b87ab","subject":"Merge \"Drop basepython\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/77829ccdb474c4755be83ef6bdf09430906b87ab"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 16:10:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 16:18:34.000000000","tz":0},"subject":"Replace obsolete PCRE packages","message":"Replace obsolete PCRE packages\n\npcre3 was removed from recent debian-based releases (eg. Trixie[1]).\n\nAlso RHEL10/CentOS Stream 10 no longer ships pcre in favor of pcre2.\n\n[1] https://lists.debian.org/debian-devel/2021/11/msg00176.html\n\nChange-Id: I7725cdb8af787bbaa387f1235d128f28eab3719f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/13a6285ec12dc343211f0ddfe418c951e39dcd84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/13a6285ec12dc343211f0ddfe418c951e39dcd84"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b9489e5294eb613627339a36dd066e8d5e103e78":{"kind":"REWORK","_number":3,"created":"2025-12-07 16:29:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/970061/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/61/970061/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/61/970061/3"}}},"commit":{"parents":[{"commit":"77829ccdb474c4755be83ef6bdf09430906b87ab","subject":"Merge \"Drop basepython\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/77829ccdb474c4755be83ef6bdf09430906b87ab"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 16:10:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 16:29:25.000000000","tz":540},"subject":"Replace obsolete PCRE3 package","message":"Replace obsolete PCRE3 package\n\nIt was replaced by PCRE2 library.\n\nChange-Id: I7725cdb8af787bbaa387f1235d128f28eab3719f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/b9489e5294eb613627339a36dd066e8d5e103e78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/b9489e5294eb613627339a36dd066e8d5e103e78"}]},"branch":"refs/heads/master"},"061b8cedde04a41d28f431d812ee0c5fc329ee14":{"kind":"REWORK","_number":4,"created":"2025-12-07 17:18:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/970061/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/61/970061/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/61/970061/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/61/970061/4"}}},"commit":{"parents":[{"commit":"77829ccdb474c4755be83ef6bdf09430906b87ab","subject":"Merge \"Drop basepython\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/77829ccdb474c4755be83ef6bdf09430906b87ab"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 16:10:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 17:18:28.000000000","tz":540},"subject":"Replace obsolete PCRE packages","message":"Replace obsolete PCRE packages\n\npcre3 was removed from recent debian-based releases (eg. Trixie[1]).\n\nAlso RHEL10/CentOS Stream 10 no longer ships pcre in favor of pcre2.\n\n[1] https://lists.debian.org/debian-devel/2021/11/msg00176.html\n\nChange-Id: I7725cdb8af787bbaa387f1235d128f28eab3719f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/061b8cedde04a41d28f431d812ee0c5fc329ee14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/061b8cedde04a41d28f431d812ee0c5fc329ee14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
