)]}'
{"id":"openstack%2Fwhereto~987675","triplet_id":"openstack%2Fwhereto~master~Iff9b3163601c8419f7491e1bfcf5293689a75e93","project":"openstack/whereto","branch":"master","topic":"deprecate-version-module","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-14 15:55:11.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Iff9b3163601c8419f7491e1bfcf5293689a75e93","subject":"Deprecate version attribute","status":"MERGED","created":"2026-05-07 14:47:28.000000000","updated":"2026-05-14 16:11:10.000000000","submitted":"2026-05-14 16:10:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987675-deprecate-version-module","meta_rev_id":"33361072ce9e05926786f4a9f58e5f44679c5fbc","_number":987675,"virtual_id_number":987675,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-14 16:10:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-12 20:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-05-14 15:55:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-07 15:00:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-12 20:34:31.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"122a2110f56bb954a6f98c907b72617b21e97001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-07 14:47:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2556684547542ba32225f686bf28d16c383b103","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-07 14:55:39.000000000","message":"Topic set to deprecate-version-module","accounts_in_message":[],"_revision_number":1},{"id":"1d7b63f8659ab7fcd84d0886bc34aa6419208a82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 15:00:35.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eae7bef658da4dd0ad14ad891cf39015\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d9aa8aa14fc44fc8a06ff882aa81181 : FAILURE in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54d1a838b2b24953a398710574145c2c : SUCCESS in 2m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b50987ac1a2142719ab3c1cd1dc081e5 : SUCCESS in 1m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/694dedf0ff5744aa9280faa0e57fbe53 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8808ab14fbba4317956996dc5151e7e8 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e930a1add8724ae91002cb19350ce2ffec6294f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-07 15:51:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d23f9da85c4f73070fe88cb5fd898f9fdaee5eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 16:01:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c55b9270f634fe39d80f6dc256ef33e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d7e9f6123c4c888a42815001ff2311 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84c214402e1c49ae85c40fa8e6fed5b3 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5e7eccd5edf430f89c42dfc862281e3 : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3b7f713e688e478b8e4bea74b29b3d30 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fa89e88e5414f3e9d4dfd8a2e4b4461 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"8a3c721c91b2d028cb1b1ea83624559db0e27372","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-12 20:34:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6d21bd33f822819b7cdb99b25edd1e23b52dc2ad","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-14 15:55:11.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ab20f8569744d4c268b264841c624d69c0c9f97a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 15:56:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"890326ee538ef9fba6dcecf650e58405a0aa7a30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 16:10:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95fb9c8476404922bb9a5fe1958bb4eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ebacaaa23b547559b110e6b1c00ad20 : SUCCESS in 6m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81d95819d74843a1a5a0e1ced8953461 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed0a2141bac4447d91c13f922d0a6bfc : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4c50ec6f5a44fc85e9310cbffe31e5 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"370ca90363b1e3e676a6db578a82525e88842ecd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 16:10:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"33361072ce9e05926786f4a9f58e5f44679c5fbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 16:11:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de173413432c42029e8bd9b8fac743b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ae760a428fc4ed18d1edfccb1606928 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9ae28723b3bf94e195728381f80c80e0e53e927f","revisions":{"603768287a6b6eab2094e82de0901b1078cb875b":{"kind":"REWORK","_number":1,"created":"2026-05-07 14:47:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/75/987675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/75/987675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/75/987675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/75/987675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/75/987675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/75/987675/1"}}},"commit":{"parents":[{"commit":"cd0502b34afbe654a4cfa7cd83924edd860b441c","subject":"ruff: Configure hacking as external linter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/cd0502b34afbe654a4cfa7cd83924edd860b441c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-07 14:47:15.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-07 14:47:21.000000000","tz":60},"subject":"Deprecate version attribute","message":"Deprecate version attribute\n\nChange-Id: Iff9b3163601c8419f7491e1bfcf5293689a75e93\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/603768287a6b6eab2094e82de0901b1078cb875b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/603768287a6b6eab2094e82de0901b1078cb875b"}]},"branch":"refs/heads/master"},"9ae28723b3bf94e195728381f80c80e0e53e927f":{"kind":"REWORK","_number":2,"created":"2026-05-07 15:51:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/75/987675/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whereto","ref":"refs/changes/75/987675/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whereto refs/changes/75/987675/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whereto refs/changes/75/987675/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whereto refs/changes/75/987675/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whereto refs/changes/75/987675/2"}}},"commit":{"parents":[{"commit":"5d45c72ba9347f3bbf73f59c3290ff656d96da55","subject":"Update packaging configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/5d45c72ba9347f3bbf73f59c3290ff656d96da55"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-07 14:47:15.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-07 15:50:54.000000000","tz":60},"subject":"Deprecate version attribute","message":"Deprecate version attribute\n\nChange-Id: Iff9b3163601c8419f7491e1bfcf5293689a75e93\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/9ae28723b3bf94e195728381f80c80e0e53e927f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whereto/commit/9ae28723b3bf94e195728381f80c80e0e53e927f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
