)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~1003958","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~If1540feafa2b5893ec080bbdddd30baa485ef494","project":"openstack/whitebox-tempest-plugin","branch":"master","attention_set":{"38744":{"account":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"last_update":"2026-09-08 16:06:09.000000000","reason":"A robot voted negatively on a label"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-08 13:48:11.000000000","reason":"\u003cGERRIT_ACCOUNT_38744\u003e replied on the change","reason_account":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If1540feafa2b5893ec080bbdddd30baa485ef494","subject":"Add Intel TDX boot tests","status":"NEW","created":"2026-09-04 13:38:34.000000000","updated":"2026-09-08 16:06:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d5657dbc91b55a5509beca3f1d1ae752a0feef87","_number":1003958,"virtual_id_number":1003958,"owner":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-08 16:06:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 13:51:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 16:21:05.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"30131c184bee3ca6529224219de759bd4807b629","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-09-04 13:38:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5cc6461106e2671eda8aa02f37a834091d50005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 13:51:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c02968f13e3248709bdfb6e076c81268\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/14ba63a52b77455fb9f5b1155e8f3532 : RETRY_LIMIT in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bf22910e1a144ab84cc833dbc2c8945 : SUCCESS in 4m 02s\n- whitebox-devstack-multinode-2026-1 https://zuul.opendev.org/t/openstack/build/cc86196d3f764170b5e781aaa088e87a : RETRY_LIMIT in 3m 21s\n- whitebox-devstack-multinode-2025-2 https://zuul.opendev.org/t/openstack/build/cd285aa635c04d18bdd90208f3c66a68 : RETRY_LIMIT in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"030829e62996f6984f5e6409cb347ddfd7fd0ded","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 16:21:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"775f6c651210d64caf5addcfb6690ae3878b0887","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-09-08 13:47:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8e794aefd9f36f157cbac2118204906d06045df7","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-09-08 13:48:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d5657dbc91b55a5509beca3f1d1ae752a0feef87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:06:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb8ebdbf59aa4388982e1170b254a415\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/8231b84b380d4dabbf047c666d6fa779 : FAILURE in 2h 14m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e6293b8f2b46f59dfee6509721b556 : SUCCESS in 2m 42s\n- whitebox-devstack-multinode-2026-1 https://zuul.opendev.org/t/openstack/build/a74168e4df984b6a9305797e522b93cc : SUCCESS in 2h 10m 44s\n- whitebox-devstack-multinode-2025-2 https://zuul.opendev.org/t/openstack/build/13f7eb3ff91141048add4684690f8ff0 : SUCCESS in 2h 09m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0def542ccd9324a751f89e265e852591343ecfff","revisions":{"ef77023c881412334395f12b1c7eeca35983b139":{"kind":"REWORK","_number":1,"created":"2026-09-04 13:38:34.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/58/1003958/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/58/1003958/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/1003958/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/1003958/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/1003958/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/1003958/1"}}},"commit":{"parents":[{"commit":"66def1bcf85ac6ea0f1c6a8583aef963c206e96c","subject":"Merge \"Set placement microversion for VGPUTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/66def1bcf85ac6ea0f1c6a8583aef963c206e96c"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-27 14:52:15.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-09-03 14:18:54.000000000","tz":-420},"subject":"Add Intel TDX boot tests","message":"Add Intel TDX boot tests\n\nAdds a new TestTDXBoot test class covering guests booted with Intel\nTDX memory encryption. Tests include with and without Secure Boot.\n\nEach test also verifies that the relevant XML is correctly generated,\nthat the machine does in fact boot with Intel TDX enabled and that the\nresource provider correctly tracks the available Intel TDX keys. For\nsecure boot tests the actual status of Secure Boot is also verified.\n\nAdds two new config options, [compute-feature-enabled]/intel_tdx and\nintel_tdx_secure_boot, to enable/skip these tests based on deployment\ncapabilities. Note that not all deployments include Intel TDX firmware\nboth with or without Secure Boot support.\n\nThis was tested with an Ubuntu 26.04 cloud image. An image used for\ntesting this features needs Intel TDX support. Additionally, it also\nneeds to support `systemd-detect-virt --cvm` and `mokutil --sb-state`.\n\nChange-Id: If1540feafa2b5893ec080bbdddd30baa485ef494\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ef77023c881412334395f12b1c7eeca35983b139"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ef77023c881412334395f12b1c7eeca35983b139"}]},"branch":"refs/heads/master"},"0def542ccd9324a751f89e265e852591343ecfff":{"kind":"REWORK","_number":2,"created":"2026-09-08 13:47:41.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/58/1003958/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/58/1003958/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/1003958/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/1003958/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/1003958/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/1003958/2"}}},"commit":{"parents":[{"commit":"66def1bcf85ac6ea0f1c6a8583aef963c206e96c","subject":"Merge \"Set placement microversion for VGPUTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/66def1bcf85ac6ea0f1c6a8583aef963c206e96c"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-27 14:52:15.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-09-08 13:45:58.000000000","tz":-420},"subject":"Add Intel TDX boot tests","message":"Add Intel TDX boot tests\n\nAdds a new TestTDXBoot test class covering guests booted with Intel\nTDX memory encryption. Tests include with and without Secure Boot.\n\nEach test also verifies that the relevant XML is correctly generated,\nthat the machine does in fact boot with Intel TDX enabled and that the\nresource provider correctly tracks the available Intel TDX keys. For\nsecure boot tests the actual status of Secure Boot is also verified.\n\nAdds two new config options, [compute-feature-enabled]/intel_tdx and\nintel_tdx_secure_boot, to enable/skip these tests based on deployment\ncapabilities. Note that not all deployments include Intel TDX firmware\nboth with or without Secure Boot support.\n\nThis was tested with an Ubuntu 26.04 cloud image. An image used for\ntesting this features needs Intel TDX support. Additionally, it also\nneeds to support `systemd-detect-virt --cvm` and `mokutil --sb-state`.\n\nChange-Id: If1540feafa2b5893ec080bbdddd30baa485ef494\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/0def542ccd9324a751f89e265e852591343ecfff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/0def542ccd9324a751f89e265e852591343ecfff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
