)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~745961","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~Idd7c29fc321502cd48a8e34cee62894457401a5e","project":"openstack/whitebox-tempest-plugin","branch":"master","hashtags":[],"change_id":"Idd7c29fc321502cd48a8e34cee62894457401a5e","subject":"Test selinux labels in QEMU/KVM environment","status":"MERGED","created":"2020-08-12 20:37:32.000000000","updated":"2020-09-17 18:55:08.000000000","submitted":"2020-09-17 18:55:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745961-1600368908742-e889b14a","meta_rev_id":"abd462b85c31a75204cffff32835cb561672c80e","_number":745961,"virtual_id_number":745961,"owner":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},{"value":0,"date":"2020-09-17 13:21:25.000000000","_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-17 18:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":2,"date":"2020-09-15 13:45:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},{"value":0,"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":1,"date":"2020-09-17 17:33:23.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},{"value":0,"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 15:34:11.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"state":"REVIEWER"},{"updated":"2020-08-13 15:35:06.000000000","updated_by":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"reviewer":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"state":"REVIEWER"},{"updated":"2020-09-14 15:48:06.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2020-09-14 17:37:47.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-09-17 17:33:23.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2020-09-17 18:55:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bdd01f72753cff38f120d041cf44661b30c783a8","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-08-12 20:37:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a1fe7b956c07bdc69e8cfdda8144304d32abf9e","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-08-12 21:34:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nA nit and a question - also, since this won\u0027t run upstream, I\u0027d like to see a downstream run passing first before merging.","accounts_in_message":[],"_revision_number":1},{"id":"644ec371390ca141e1a1f55b4c148a487f3218b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 22:07:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcbfcd41364e40fcbb8ebeb57dc2fc87 : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3ebd4a8bf6e44da8822e5c44526cd09 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d6335517c0a493e9e594e37a5c6595a : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/527c1ef092884518a95affa9cb45960b : SUCCESS in 4m 15s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/bfaeed57820e465c80db1f18e0308e00 : SUCCESS in 1h 10m 03s\n- whitebox-multinode-multinuma-devstack https://zuul.opendev.org/t/openstack/build/6675f7e8c87140b8bf2e3298f32125e9 : RETRY_LIMIT in 8m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"051209862ba1a73028ee6f01f65dc2cee9bf027c","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-08-12 22:07:35.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nI can create another patch for [DNM] testing, if that looks good??","accounts_in_message":[],"_revision_number":1},{"id":"576eb54122c1277bbea4c7e7a5e446abc10129c8","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-08-13 15:33:01.000000000","message":"Patch Set 1:\n\nTesting it here: https://review.opendev.org/746162","accounts_in_message":[],"_revision_number":1},{"id":"baa8f1ea465aaf9981f091338dc0e65c57f834a5","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-08-13 16:32:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e910133969368336774ed678b2374e77ec6ef77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 17:21:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06e1dd9b70794065b2cd10ee874e0d46 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fec90630c7d45a8871105d54722ea88 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fe29f777822414182c61ed3e62c04d6 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f7b8e9005b74aea9df7429c47ca9732 : SUCCESS in 4m 25s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/d4174efc6b50413d851f67856c87a0db : FAILURE in 44m 44s\n- whitebox-multinode-multinuma-devstack https://zuul.opendev.org/t/openstack/build/b575b4e1800c40649ca941dec9c6a4c7 : RETRY_LIMIT in 9m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57802d46dcbd7acd9ae312fcabeb6e647546f586","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-08-13 19:07:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"29b305697436fee5e3a17cd82ed95e8bd809efd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 20:08:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9eb768c488c4536898192e884c19713 : SUCCESS in 2m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/745f558959e34befa637443663d99922 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb2b84e691ff4464ba5a94fb63af26eb : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bf9faeb12e64edb8de624cf31bc2470 : SUCCESS in 4m 34s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/7b95d6d06a174521b1c05b0121789b10 : SUCCESS in 59m 31s\n- whitebox-multinode-multinuma-devstack https://zuul.opendev.org/t/openstack/build/37f41af66e8d4b78aec37861001c07bc : RETRY_LIMIT in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b06bb1ea909f05f4248ec3d69debfdafcb41f97","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-18 16:58:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ecc8ca5fd3c6d8eb1f2b9c6149f3d626af8448fa","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-18 17:57:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1409cf53f40f29a63093fcbcf10e0c49aff778c2","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-08-19 17:51:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"adfd4908fd7c9c892495a7c8dd414eb22e11a845","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-08-20 15:09:42.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1361f4b42b89aafd08c68cae118fa9acf4d5e085","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-08-20 19:16:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2dcb26b6e1d53bd73e206e041170a7a555469fbc","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-08-20 19:41:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a67d90ef6ee907f927f63cf306c1a99a21a3f844","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-08-20 19:42:47.000000000","message":"Patch Set 4:\n\ndownstream test results:\n\n(overcloud) [stack@undercloud-0 tempest-dir]$ tempest run --regex SelinuxLabelsTest\n{0} whitebox_tempest_plugin.api.compute.test_validate_selinux_labels.SelinuxLabelsTest.test_create_server_with_label_check [7.467161s] ... ok\n{0} whitebox_tempest_plugin.api.compute.test_validate_selinux_labels.SelinuxLabelsTest.test_live_migrate_with_label_check [13.859682s] ... ok\n{0} whitebox_tempest_plugin.api.compute.test_validate_selinux_labels.SelinuxLabelsTest.test_resize_with_label_check [64.174614s] ... ok\n\n\u003d\u003d\u003d\u003d\u003d\u003d\nTotals\n\u003d\u003d\u003d\u003d\u003d\u003d\nRan: 3 tests in 85.5036 sec.\n - Passed: 3\n - Skipped: 0\n - Expected Fail: 0\n - Unexpected Success: 0\n - Failed: 0\nSum of execute time for each test: 85.5015 sec.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nWorker Balance\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n - Worker 0 (3 tests) \u003d\u003e 0:01:25.503610","accounts_in_message":[],"_revision_number":4},{"id":"9edda60661997c95f997738bbce706dfdc797e06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:35:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a12e2cfeb24654a54b64f7e68f177f : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c14a7173de904ff5bce84fc4fc9674a5 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d19ceb11b2e545bd8e6c16c53ef1ffbf : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/396f219e485e49b6a66bfb3ba45731ec : SUCCESS in 4m 22s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/7b029cd981a84e8eb460bcb3363896bd : SUCCESS in 51m 38s\n- whitebox-multinode-multinuma-devstack https://zuul.opendev.org/t/openstack/build/3ddac829bde94d95b8c1402f842b31a2 : RETRY_LIMIT in 10m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8287da177d0413ae4c83b98372b31376b34c3600","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-08-24 18:26:28.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)\n\nLooking good, just an accumulation of nits. I\u0027d also like Kashyap\u0027s signoff on this before merging.","accounts_in_message":[],"_revision_number":4},{"id":"bbd9e6bce6f2ddc1c985ff2649c3d8ea813aa336","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-08-26 14:30:25.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"24c77bab62875ea33f67cd1fd69e431195159dac","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-08-26 14:35:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8c055d17bb6c1262b54dabbb79c343e2dff0431d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 15:38:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f587791999d433c9c2c21231724175b : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45dade6ea923491d869c4dfbd4d43ad5 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a3cf8b0af034f93a40cad64a7fb649e : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c58cf10d7547476db95f155e56f5ae03 : SUCCESS in 4m 47s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/fd8839f96e624e3a913da8ab973e9f04 : SUCCESS in 55m 12s\n- whitebox-multinode-multinuma-devstack https://zuul.opendev.org/t/openstack/build/fed33dbb49ac435f9f27c1d8f6814410 : RETRY_LIMIT in 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"923ca28bef24bd52c2c8081531622e53832ca84e","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-09-14 15:48:06.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nI see that Artom and Sean have done a bunch of review rounds\n\nThe core test itself is validating the right XML bits.  Thanks!","accounts_in_message":[],"_revision_number":5},{"id":"81c40a11d8be13c447f12668e7c117c915e9265b","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-09-14 16:18:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"12f9803a90a25a8fa20962791f1d9c20f1cbf4b8","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-09-14 16:20:13.000000000","message":"Patch Set 5:\n\nlet me know artom/sean or anyone else have any other review. Otherwise I will rebase this patch","accounts_in_message":[],"_revision_number":5},{"id":"8966bea060159d7206d0e3b3985ceeaed840e692","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-09-14 16:41:28.000000000","message":"Patch Set 5:\n\n\u003e let me know artom/sean or anyone else have any other review.\n \u003e Otherwise I will rebase this patch\n\nI think you\u0027re good. Rebase and I\u0027ll +A","accounts_in_message":[],"_revision_number":5},{"id":"37d6f20e7a4227af1b569a5ff1f3c08ed1ec5308","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-09-14 16:58:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d8d2b41f126509374bad7d9348ba4f4092991311","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-09-14 17:37:47.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)\n\nim going to proxy kashyaps +1 asserting that the element that need to be checked are check.\n\ni was not aware we shuld also check the imagelabel but if kasyap says we shoudl check both then im ok with that.\n\noverall this looks ok to","accounts_in_message":[],"_revision_number":6},{"id":"b7ff5c084542efde0a45c6fa5d9857f4dcb1cfd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 18:12:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5a553fd68294d7fa785589b1832ed6b : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe27b5b1994f4c2fa0bad4791bd7d246 : FAILURE in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5dee348dfa0431db8e7043e0a0a2911 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f962b50630949eaabb4fd387eae3021 : SUCCESS in 5m 00s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/96161e31d82046f1b9b0c8dd64d68ed0 : FAILURE in 1h 10m 37s\n- whitebox-multinode-multinuma-devstack https://zuul.opendev.org/t/openstack/build/c70b62ebfa404382ad950808aaee475d : RETRY_LIMIT in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8be58c0055a377a5524eadf44ce55f736a0241e5","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-09-14 20:05:44.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a15d990e72c6229354ed5273ab785659158313f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 21:11:44.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09b79d9099ea4529a0e55085d5ed8e3d : SUCCESS in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ae89ac485f94057a19e5c7b82923094 : FAILURE in 2m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35d651813c9b4be29052dae8518def10 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee5338fb16cf47c6b5f212048ddb9c5a : SUCCESS in 3m 30s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/09efc699455449d3a34d854639fd908c : SUCCESS in 1h 04m 14s\n- whitebox-multinode-multinuma-devstack https://zuul.opendev.org/t/openstack/build/d20453deea9d4f90b3802f9dcde42571 : RETRY_LIMIT in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dd32005eeb9388b6478d4fcd08e9f86fc8efbc8e","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-09-15 12:56:39.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"0163873dd96b2e83e78b8729fdb9bc87e8e000f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 13:33:21.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b312ffa370f84e6a9752e90778531c67 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b228df6bc5f945a082cf79b87f9880db : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1c558ee56e84c4caa62a302db7a38c9 : SUCCESS in 4m 02s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/5cedb8a2fffe4ae6ab130ccfe9cc8f6d : FAILURE in 14m 24s\n- whitebox-multinode-multinuma-devstack https://zuul.opendev.org/t/openstack/build/8e1fa14e29a34caa8af3c4f4594ffcac : RETRY_LIMIT in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"332b3129dc4a4a6a0deb23b6f302d761232d25e3","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-09-15 13:38:14.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5b462dc60fa6c8833142fc2f48c2b7750eb4bff5","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-09-15 13:45:04.000000000","message":"Patch Set 6:\n\nYou need to rebase I think","accounts_in_message":[],"_revision_number":6},{"id":"b84ea64621129740988f11dd50d93947a6c4c9a0","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-09-15 13:45:08.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"6468ecf5065e0912ac2f57e711952f33d4a13418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 14:17:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0daf8de37d6e418998c933fa88460604 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d2b65dd148b4bc2b6d0b36fecae2bf5 : POST_FAILURE in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb54a37a49084c75a908155db6a17554 : SUCCESS in 5m 56s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/dd315e55eca14602aafec85cb4e31eb5 : FAILURE in 14m 30s\n- whitebox-multinode-multinuma-devstack https://zuul.opendev.org/t/openstack/build/c9ff2cd98e6f457290f3e6e339e2ed07 : RETRY_LIMIT in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fab17c5a39a0c0550d6ed68764b53e87a7449ec5","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-09-17 13:21:25.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"a9b58e87e9aa652626e821180de96c0ba8693d6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 14:37:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/603982ce01a4415a9cd7f98dfbb1647d : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15350ff5ae5742719398978a9161c2b6 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f00622780a645218d7e1e154dd39d0c : SUCCESS in 5m 31s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/21a7c43e5d834b26bf8819ee35ad76c5 : SUCCESS in 1h 12m 22s\n- whitebox-multinode-multinuma-devstack https://zuul.opendev.org/t/openstack/build/84c54bc17d4d4047b459802d1e77f1aa : RETRY_LIMIT in 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1a1eb7dc31cae27ee9db7ee101a1de1b08acef1e","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-09-17 17:33:23.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e987f06e6c47fc47cd489ed1846e2fc723a3e700","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 17:37:44.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9677a890a1600f93c16758cb99e292ab50a85508","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 18:55:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"abd462b85c31a75204cffff32835cb561672c80e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 18:55:08.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/026232c042304a8d8490c67baea4516f : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a75193aa12494325b344fe27c6b341e4 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25ab70dd6f6f4c84860078be8eea92c8 : SUCCESS in 5m 33s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/44b068fb623f441fa75a04d882aa544e : SUCCESS in 1h 14m 40s\n- whitebox-multinode-multinuma-devstack https://zuul.opendev.org/t/openstack/build/6b67a83942a14bdb98196d3f9d7827ff : RETRY_LIMIT in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"82c9e79d5219ac4b3aedd0bafecb9013e4cbca33","revisions":{"6852bf0a26113c27d532176b23aa658ecefb992c":{"kind":"REWORK","_number":1,"created":"2020-08-12 20:37:32.000000000","uploader":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"ref":"refs/changes/61/745961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/61/745961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/1"}}},"commit":{"parents":[{"commit":"97713d8bf73a33faea3a2f629b44764773e8656c","subject":"New [whitebox-hardware]/cpu_topology option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/97713d8bf73a33faea3a2f629b44764773e8656c"}]}],"author":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-12 20:31:48.000000000","tz":-240},"committer":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-12 20:31:48.000000000","tz":-240},"subject":"Test selinux labels in QEMU/KVM environment","message":"Test selinux labels in QEMU/KVM environment\n\nThis will test the selinux labels for the deployed\ninstance, resize instance and live migrated instance\nusing the xml file.\n\nChange-Id: Idd7c29fc321502cd48a8e34cee62894457401a5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6852bf0a26113c27d532176b23aa658ecefb992c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6852bf0a26113c27d532176b23aa658ecefb992c"}]},"branch":"refs/heads/master"},"8075b2822c7b2a00e5c6f77e64aef644acb4b8b2":{"kind":"REWORK","_number":2,"created":"2020-08-13 16:32:55.000000000","uploader":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"ref":"refs/changes/61/745961/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/61/745961/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/2"}}},"commit":{"parents":[{"commit":"97713d8bf73a33faea3a2f629b44764773e8656c","subject":"New [whitebox-hardware]/cpu_topology option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/97713d8bf73a33faea3a2f629b44764773e8656c"}]}],"author":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-12 20:31:48.000000000","tz":-240},"committer":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-13 16:32:47.000000000","tz":-240},"subject":"Test selinux labels in QEMU/KVM environment","message":"Test selinux labels in QEMU/KVM environment\n\nThis will test the selinux labels for the deployed\ninstance, resize instance and live migrated instance\nusing the xml file.\n\nChange-Id: Idd7c29fc321502cd48a8e34cee62894457401a5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/8075b2822c7b2a00e5c6f77e64aef644acb4b8b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/8075b2822c7b2a00e5c6f77e64aef644acb4b8b2"}]},"branch":"refs/heads/master"},"0fa5221c63b3e4843937419f8edd249578111749":{"kind":"REWORK","_number":3,"created":"2020-08-20 19:16:04.000000000","uploader":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"ref":"refs/changes/61/745961/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/61/745961/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/3"}}},"commit":{"parents":[{"commit":"97713d8bf73a33faea3a2f629b44764773e8656c","subject":"New [whitebox-hardware]/cpu_topology option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/97713d8bf73a33faea3a2f629b44764773e8656c"}]}],"author":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-12 20:31:48.000000000","tz":-240},"committer":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-20 19:15:52.000000000","tz":-240},"subject":"Test selinux labels in QEMU/KVM environment","message":"Test selinux labels in QEMU/KVM environment\n\nThis will test the selinux labels for the deployed\ninstance, resize instance and live migrated instance\nusing the xml file.\n\nChange-Id: Idd7c29fc321502cd48a8e34cee62894457401a5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/0fa5221c63b3e4843937419f8edd249578111749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/0fa5221c63b3e4843937419f8edd249578111749"}]},"branch":"refs/heads/master"},"239f3070a2fe5384a1ac8e85a7a3b542b430d3a4":{"kind":"REWORK","_number":4,"created":"2020-08-20 19:41:40.000000000","uploader":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"ref":"refs/changes/61/745961/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/61/745961/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/4"}}},"commit":{"parents":[{"commit":"97713d8bf73a33faea3a2f629b44764773e8656c","subject":"New [whitebox-hardware]/cpu_topology option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/97713d8bf73a33faea3a2f629b44764773e8656c"}]}],"author":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-12 20:31:48.000000000","tz":-240},"committer":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-20 19:41:32.000000000","tz":-240},"subject":"Test selinux labels in QEMU/KVM environment","message":"Test selinux labels in QEMU/KVM environment\n\nThis will test the selinux labels for the deployed\ninstance, resize instance and live migrated instance\nusing the xml file.\n\nChange-Id: Idd7c29fc321502cd48a8e34cee62894457401a5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/239f3070a2fe5384a1ac8e85a7a3b542b430d3a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/239f3070a2fe5384a1ac8e85a7a3b542b430d3a4"}]},"branch":"refs/heads/master"},"6fbd7f281e796dcbcce09bfffc83f17361b76241":{"kind":"REWORK","_number":5,"created":"2020-08-26 14:35:13.000000000","uploader":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"ref":"refs/changes/61/745961/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/61/745961/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/5"}}},"commit":{"parents":[{"commit":"97713d8bf73a33faea3a2f629b44764773e8656c","subject":"New [whitebox-hardware]/cpu_topology option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/97713d8bf73a33faea3a2f629b44764773e8656c"}]}],"author":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-12 20:31:48.000000000","tz":-240},"committer":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-26 14:35:02.000000000","tz":-240},"subject":"Test selinux labels in QEMU/KVM environment","message":"Test selinux labels in QEMU/KVM environment\n\nThis will test the selinux labels for the deployed\ninstance, resize instance and live migrated instance\nusing the xml file.\n\nChange-Id: Idd7c29fc321502cd48a8e34cee62894457401a5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6fbd7f281e796dcbcce09bfffc83f17361b76241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6fbd7f281e796dcbcce09bfffc83f17361b76241"}]},"branch":"refs/heads/master"},"afa3af8a25f8eb4d7a0b1175eb2205cb9944e21b":{"kind":"REWORK","_number":6,"created":"2020-09-14 16:58:23.000000000","uploader":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"ref":"refs/changes/61/745961/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/61/745961/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/6"}}},"commit":{"parents":[{"commit":"f4e173604a47978992cee0ffc04fe1310a803ff5","subject":"Add paramiko version atleast 2.7.0 in requirement.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/f4e173604a47978992cee0ffc04fe1310a803ff5"}]}],"author":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-12 20:31:48.000000000","tz":-240},"committer":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-09-14 16:58:11.000000000","tz":-240},"subject":"Test selinux labels in QEMU/KVM environment","message":"Test selinux labels in QEMU/KVM environment\n\nThis will test the selinux labels for the deployed\ninstance, resize instance and live migrated instance\nusing the xml file.\n\nChange-Id: Idd7c29fc321502cd48a8e34cee62894457401a5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/afa3af8a25f8eb4d7a0b1175eb2205cb9944e21b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/afa3af8a25f8eb4d7a0b1175eb2205cb9944e21b"}]},"branch":"refs/heads/master"},"82c9e79d5219ac4b3aedd0bafecb9013e4cbca33":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-15 13:45:08.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/61/745961/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/61/745961/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/61/745961/7"}}},"commit":{"parents":[{"commit":"c069fde0b5333ec259384f2618ed3ead3e76e92e","subject":"Get rid of py27 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/c069fde0b5333ec259384f2618ed3ead3e76e92e"}]}],"author":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-08-12 20:31:48.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"notartom@gmail.com","date":"2020-09-15 13:45:08.000000000","tz":0},"subject":"Test selinux labels in QEMU/KVM environment","message":"Test selinux labels in QEMU/KVM environment\n\nThis will test the selinux labels for the deployed\ninstance, resize instance and live migrated instance\nusing the xml file.\n\nChange-Id: Idd7c29fc321502cd48a8e34cee62894457401a5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/82c9e79d5219ac4b3aedd0bafecb9013e4cbca33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/82c9e79d5219ac4b3aedd0bafecb9013e4cbca33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
