)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~767185","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~I8c597e7fbf40cef046cf401661697ec0b5fc1101","project":"openstack/whitebox-tempest-plugin","branch":"master","topic":"remove-max-compute-nodes","hashtags":[],"change_id":"I8c597e7fbf40cef046cf401661697ec0b5fc1101","subject":"Make test_cpu_dedicated use the same_host scheduler hint","status":"ABANDONED","created":"2020-12-15 17:02:20.000000000","updated":"2021-12-06 20:44:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f75fc34a7d77ab7c1d4152e562396122bdfa766c","_number":767185,"virtual_id_number":767185,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-16 02:27:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"all":[{"tag":"autogenerated:zuul:automatic-ci","value":-1,"date":"2020-12-16 02:01:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 17:02:35.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2020-12-15 17:42:21.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"REVIEWER"},{"updated":"2020-12-15 17:52:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41a80c53966bd557c3e3a3a008e215bb34823849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-12-15 17:02:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34e9a999cf18eb61a58b4a3e86b21fbc778523c0","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2020-12-15 17:02:35.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7efd4815662c1015d4a7f4f91488028179aaa9d1","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2020-12-15 17:42:21.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (automatic-ci pipeline). To rerun just this ci comment\n\"seans-automatic-ci: recheck\". To rerun other ci jobs comment \"recheck\".\nNote this ci does run on \"recheck\".\n\n\n- whitebox-multinode-multinuma-devstack https://zuul.seanmooney.info/t/openstack/build/cfc33f249efb4ae7aaa70da81a6ef9bd : FAILURE in 39m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3106b25688d0783b7a3a5b1ba2a0089fea5e2ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 17:52:26.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af2348e040d547e787390b2ad74d7fbd : FAILURE in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/352a9fb3e47841e8a5da4b1e1f01af81 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5f4854226a44a6a8ed50f01e1b94e75 : SUCCESS in 3m 16s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/52265430cc0b4e41990ab87b14042a4b : FAILURE in 46m 08s","accounts_in_message":[],"_revision_number":1},{"id":"133c1b93b1705fae01802041e00aa28771746c42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-12-16 00:45:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2c5e1c0ff57509dfbe903075a3343752f1a96db","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2020-12-16 00:45:33.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d008f8261f4fc7fdd9dd566e822099a76084ec7c","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2020-12-16 02:01:09.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (automatic-ci pipeline). To rerun just this ci comment\n\"seans-automatic-ci: recheck\". To rerun other ci jobs comment \"recheck\".\nNote this ci does run on \"recheck\".\n\n\n- whitebox-multinode-multinuma-devstack https://zuul.seanmooney.info/t/openstack/build/c0e63ec3166a436d98664eda5894ca01 : FAILURE in 1h 14m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3c8e37b7a38da5c632f93bb4267f6098c63f588e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 02:27:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f061d5a236c94c9ea202c3671b92e245 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0afb2f097ce4c70999ef477309293e8 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2146e16451db4df9a6614547b012dd80 : SUCCESS in 4m 29s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/ecd2a5f77ea64d2aa9e8b036384ddf94 : SUCCESS in 1h 33m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f75fc34a7d77ab7c1d4152e562396122bdfa766c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-12-06 20:44:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0cdf85ab1717a5cb5345971279ca84109bab5aa6","revisions":{"10dee48a785bd375313fc89c58309103de16428f":{"kind":"REWORK","_number":1,"created":"2020-12-15 17:02:20.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/767185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/767185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/767185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/767185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/767185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/767185/1"}}},"commit":{"parents":[{"commit":"4a4d0c6d639681c8224dbed3508acad1748f4d26","subject":"Remove restart_command config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4a4d0c6d639681c8224dbed3508acad1748f4d26"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-15 16:57:21.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-15 17:01:52.000000000","tz":-300},"subject":"Make test_cpu_dedicated use the same_host scheduler hint","message":"Make test_cpu_dedicated use the same_host scheduler hint\n\nPreviously, test_cpu_dedicated relied on having a single host in the\ndeployment to force both its instances on the same compute host. To\nthat end, it used the max_compute_nodes config option. This patches\nmakes it use the same_host scheduler hint.\n\nChange-Id: I8c597e7fbf40cef046cf401661697ec0b5fc1101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/10dee48a785bd375313fc89c58309103de16428f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/10dee48a785bd375313fc89c58309103de16428f"}]},"branch":"refs/heads/master"},"0cdf85ab1717a5cb5345971279ca84109bab5aa6":{"kind":"REWORK","_number":2,"created":"2020-12-16 00:45:18.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/767185/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/767185/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/767185/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/767185/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/767185/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/767185/2"}}},"commit":{"parents":[{"commit":"4a4d0c6d639681c8224dbed3508acad1748f4d26","subject":"Remove restart_command config option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4a4d0c6d639681c8224dbed3508acad1748f4d26"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-15 16:57:21.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-12-16 00:45:11.000000000","tz":-300},"subject":"Make test_cpu_dedicated use the same_host scheduler hint","message":"Make test_cpu_dedicated use the same_host scheduler hint\n\nPreviously, test_cpu_dedicated relied on having a single host in the\ndeployment to force both its instances on the same compute host. To\nthat end, it used the max_compute_nodes config option. This patches\nmakes it use the same_host scheduler hint.\n\nChange-Id: I8c597e7fbf40cef046cf401661697ec0b5fc1101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/0cdf85ab1717a5cb5345971279ca84109bab5aa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/0cdf85ab1717a5cb5345971279ca84109bab5aa6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
