)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~805297","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~Ibc8a624153523aa14e073c46df2a5c793c5c230f","project":"openstack/whitebox-tempest-plugin","branch":"master","topic":"interface_change_rx_queue","hashtags":[],"change_id":"Ibc8a624153523aa14e073c46df2a5c793c5c230f","subject":"Simplify test_rx_tx_queue_size","status":"MERGED","created":"2021-08-19 20:30:12.000000000","updated":"2021-08-28 05:29:25.000000000","submitted":"2021-08-28 05:29:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805297-interface_change_rx_queue","meta_rev_id":"9a4f4836b8e962c0437437fbe6a305f0d86ba63b","_number":805297,"virtual_id_number":805297,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-28 05:29:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":2,"date":"2021-08-28 05:29:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":1,"date":"2021-08-27 22:38:59.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 21:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-25 20:44:56.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"state":"REVIEWER"}],"messages":[{"id":"77e5ae0860bf8916991cb38e79e3685621dc9120","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-19 20:30:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b2edc457d4e936687b77cfae69e1022a99da6d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 21:15:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19b94f1234f439db4940fd5b1b0870c : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7be3e5c48594493b01a2c52de6e79a8 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f496888e17cf469fb59c7e8fc226a631 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9080ac2632c04a27ad111ae2ada8298c : SUCCESS in 4m 18s (non-voting)\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/b7e21f09eb234441a635e2dcf51f6e30 : FAILURE in 41m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2200010dbf7267b67e25ae5bb132ca75cc498a27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-19 22:54:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a19493d098f3f25f58001f34c451b1a827b53662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 00:10:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cab4b1166a364143854bc12e37b36ba2 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dde1ad731c8744268b7a7be8fb537207 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54ef68d016e64c6284fdf6a76551509f : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4225b5a8783448da92d24505c9a3eb0 : SUCCESS in 4m 54s (non-voting)\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/74e852ea713446b5b8012953076d44ac : FAILURE in 1h 13m 40s","accounts_in_message":[],"_revision_number":2},{"id":"1ff3276a512ff0336014e89016aa2554908ca435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-20 00:19:26.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ac1954a693df1c2a122ca56ec89cd0e4a68d9e81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-20 01:46:25.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"daf46877f7f881eff1dfed44838c52faba6148df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 03:37:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1451910e98cc4373a06d30188794f3b7 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecf97662f277476ab932483dc8f8e420 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b36a59c9ffb44d2a8ef2d37f66aad70 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a5f340477dc4cdfa4019a57674d4db0 : SUCCESS in 4m 17s (non-voting)\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/43928e3a3af24fc489c5c330ed6d3fe0 : SUCCESS in 1h 48m 59s","accounts_in_message":[],"_revision_number":4},{"id":"1a1a045c3e32d8ad613b77e2ba02414e5b138337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-25 01:43:45.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"1a4abd41b3e3749e6a4b9338fd0b4d2c98f12ef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 03:21:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0692effe0d49474aa2d02cb19f2f9d48 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cb1347a6fec4db2a02fb88fc0085c47 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e016215420c431d88c7bc126fcec9f4 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/148bc806dfef4be790603dd60c8de0c5 : SUCCESS in 3m 58s (non-voting)\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/21f830ebdcb04df6bb7bc4610c8e8710 : FAILURE in 1h 34m 33s","accounts_in_message":[],"_revision_number":5},{"id":"8e5382802b0990cdd39a3c10f3c46fb755b63fd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-25 16:51:43.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"e33e6d330c16eb452f2a6c30aa98c0f1da67b26e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 18:33:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5386e59861c04e5095c4689243c9287c : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a5265e162c24951b6ee661a897ae896 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef36226b1e2d405f88333d7b51e745af : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bff0e4e456ae49d39b2eba76b211ff4b : SUCCESS in 5m 48s (non-voting)\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/a09fdf17dae94b1492737666e101f4bd : SUCCESS in 1h 38m 04s","accounts_in_message":[],"_revision_number":6},{"id":"0f2ac87b186dceddd519337b1b2d497a97f75aa8","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2021-08-25 20:44:56.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"1658ec606814f2e214543b50bd786ebf8382cea9","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2021-08-27 14:50:05.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"225e1f2b191a7ad096109cd0192cef94bfeaac97","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-27 15:33:47.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"aa40df27318669c27627b8d0acdcad9ecf51147d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 15:34:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7f53e8124668d6ea0dcebc6d053a574eb73b380a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 17:43:14.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"cb0cfa8a96798ba3c925c80184c4342204595846","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-27 19:51:51.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"841d55892de25b39c251f59b580d4374f6970431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 22:02:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1904064c4ee1413a970c2ce126f20225 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e0116cb545148b1941eb6e79e32b512 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fac79ca31374b11ab0b86aa7efa3a5e : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d060df38ee80451b999e0b7808488538 : SUCCESS in 5m 53s (non-voting)\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/242d11d0f79c4cbe88d1332a7bb1584a : SUCCESS in 2h 08m 07s","accounts_in_message":[],"_revision_number":6},{"id":"97d5e7f3fb83748decee719b6288920715ae8e5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 22:02:35.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3392239d7c239ce68a20621758be21c4c2700186","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-27 22:38:09.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"935fc6684a2127ce5d6cc7dd9896c43583e7af5c","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-27 22:38:59.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"37263bf9c27790426629cd73d96406fa60a5c282","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 00:14:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/596bc7f3830449c2806b68ef6802b7d1 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd032e010063492389053a6e42edc8da : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6594c48fd924a00926b0d08492740b7 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2955087ae4c94e5e8072f48d01d5f686 : SUCCESS in 4m 09s (non-voting)\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/b17146c0dedd4303a69f3d05b33887c2 : SUCCESS in 1h 33m 14s","accounts_in_message":[],"_revision_number":7},{"id":"05af58c88685a603dba80067df44124c423e8f1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 00:14:43.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1f2fbd84fac13efdf42aa398ad700688fcfd342a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 01:50:00.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b5dd722a7f4f1482cb25c20bd44973 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94ff04ad2161412e9ee25811a86afd8e : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e9bad6145bf4c0192d2ef3b2c8a4cc6 : SUCCESS in 4m 07s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/6541dd156cf74e57a436536c5d0063d5 : FAILURE in 1h 34m 04s","accounts_in_message":[],"_revision_number":7},{"id":"03d90769c0063bc0f4ebaff46e959392a8e0a4b4","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-28 01:54:58.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"aa4a75202bd16a879e3b7c3d232321a83d1e0d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 03:42:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a4251f77d68449bbdb711f6a8c850bf : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74b0ba8b675d4a108c450a9078e96172 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee8419a9bc3d468cbc4953e7ce76a818 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ddfd19ac73e4aa8862e5a1534970ba3 : SUCCESS in 4m 00s (non-voting)\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/db1feba6d1e04c5887251e9ec10fe3a8 : SUCCESS in 1h 45m 47s","accounts_in_message":[],"_revision_number":7},{"id":"7272e6118c5f7d7adcad8a5d6b5f1415e0a52a53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 03:43:17.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e5152db744727bf48011e9f52f52e79719faffd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 05:29:25.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4b91dcf4bd044ba8361fc0d84c5dfee : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eed94fcb8944d15be25c20552405906 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f3ef0c10cc94f0ba9bb24e53bbbb0f7 : SUCCESS in 3m 40s\n- whitebox-multinode-devstack https://zuul.opendev.org/t/openstack/build/2ebe424bbeed43bebad7619e6a11fd65 : SUCCESS in 1h 44m 53s","accounts_in_message":[],"_revision_number":7},{"id":"c80c91640cdd6eae82ad08693263f0eddba9da87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 05:29:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8431ed7405a585d4e7b5091a0c5b240888747175","revisions":{"4ddee2abe24165476cc3bd942181f7ab5c16c726":{"kind":"REWORK","_number":1,"created":"2021-08-19 20:30:12.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/97/805297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/97/805297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/1"}}},"commit":{"parents":[{"commit":"87ff1e441ca4c25e30ed670f9a51b39746b2c6f9","subject":"Simplify test_pointer_device_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/87ff1e441ca4c25e30ed670f9a51b39746b2c6f9"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-19 19:40:32.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-19 20:29:51.000000000","tz":-240},"subject":"Simplify test_rx_tx_queue_size","message":"Simplify test_rx_tx_queue_size\n\nDepend on tempest to create the network and subnet, and inline the\nserver creation to use the defaults and to remove the resource_setup()\n\nChange-Id: Ibc8a624153523aa14e073c46df2a5c793c5c230f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4ddee2abe24165476cc3bd942181f7ab5c16c726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4ddee2abe24165476cc3bd942181f7ab5c16c726"}]},"branch":"refs/heads/master"},"28eaf206253a9cea0022905df43d6185ba4608c1":{"kind":"REWORK","_number":2,"created":"2021-08-19 22:54:08.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/97/805297/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/97/805297/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/2"}}},"commit":{"parents":[{"commit":"464a6db4b8830db70b480cd7d8d401d5a082cee7","subject":"Simplify test_pointer_device_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/464a6db4b8830db70b480cd7d8d401d5a082cee7"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-19 19:40:32.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-19 22:54:02.000000000","tz":-240},"subject":"Simplify test_rx_tx_queue_size","message":"Simplify test_rx_tx_queue_size\n\nDepend on tempest to create the network and subnet, and inline the\nserver creation to use the defaults and to remove the resource_setup()\n\nChange-Id: Ibc8a624153523aa14e073c46df2a5c793c5c230f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/28eaf206253a9cea0022905df43d6185ba4608c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/28eaf206253a9cea0022905df43d6185ba4608c1"}]},"branch":"refs/heads/master"},"ea3dbeb1d11bd2bf48fff243b935c8969492afdf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-08-20 00:19:26.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/97/805297/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/97/805297/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/3"}}},"commit":{"parents":[{"commit":"93e91b17bc1c348df27fddbffaf0cbe0fdf24717","subject":"Simplify test_pointer_device_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/93e91b17bc1c348df27fddbffaf0cbe0fdf24717"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-19 19:40:32.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-20 00:19:20.000000000","tz":-240},"subject":"Simplify test_rx_tx_queue_size","message":"Simplify test_rx_tx_queue_size\n\nDepend on tempest to create the network and subnet, and inline the\nserver creation to use the defaults and to remove the resource_setup()\n\nChange-Id: Ibc8a624153523aa14e073c46df2a5c793c5c230f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ea3dbeb1d11bd2bf48fff243b935c8969492afdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ea3dbeb1d11bd2bf48fff243b935c8969492afdf"}]},"branch":"refs/heads/master"},"6ded240e86a0cad5da09385ea0d9ca37529119b5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-08-20 01:46:25.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/97/805297/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/97/805297/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/4"}}},"commit":{"parents":[{"commit":"ee3a9b602be5daa950d0b69ccfb10157cf8617f2","subject":"Simplify test_pointer_device_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ee3a9b602be5daa950d0b69ccfb10157cf8617f2"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-19 19:40:32.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-20 01:46:17.000000000","tz":-240},"subject":"Simplify test_rx_tx_queue_size","message":"Simplify test_rx_tx_queue_size\n\nDepend on tempest to create the network and subnet, and inline the\nserver creation to use the defaults and to remove the resource_setup()\n\nChange-Id: Ibc8a624153523aa14e073c46df2a5c793c5c230f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6ded240e86a0cad5da09385ea0d9ca37529119b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6ded240e86a0cad5da09385ea0d9ca37529119b5"}]},"branch":"refs/heads/master"},"7a619538556daf30dec87328ad6ec5606591abd2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-25 01:43:45.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/97/805297/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/97/805297/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/5"}}},"commit":{"parents":[{"commit":"d24665bbd38041048715b5f9323562650b783a66","subject":"Simplify test_pointer_device_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d24665bbd38041048715b5f9323562650b783a66"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-19 19:40:32.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-25 01:42:45.000000000","tz":-240},"subject":"Simplify test_rx_tx_queue_size","message":"Simplify test_rx_tx_queue_size\n\nDepend on tempest to create the network and subnet, and inline the\nserver creation to use the defaults and to remove the resource_setup()\n\nChange-Id: Ibc8a624153523aa14e073c46df2a5c793c5c230f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/7a619538556daf30dec87328ad6ec5606591abd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/7a619538556daf30dec87328ad6ec5606591abd2"}]},"branch":"refs/heads/master"},"4a7a35b3dc00ec216c4ef53f0cf587fa1cf1e0f7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-08-25 16:51:43.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/97/805297/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/97/805297/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/6"}}},"commit":{"parents":[{"commit":"1b654bd70ad21296e935e4befad277fa517c0a28","subject":"Simplify test_pointer_device_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/1b654bd70ad21296e935e4befad277fa517c0a28"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-19 19:40:32.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-25 16:46:36.000000000","tz":-240},"subject":"Simplify test_rx_tx_queue_size","message":"Simplify test_rx_tx_queue_size\n\nDepend on tempest to create the network and subnet, and inline the\nserver creation to use the defaults and to remove the resource_setup()\n\nChange-Id: Ibc8a624153523aa14e073c46df2a5c793c5c230f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4a7a35b3dc00ec216c4ef53f0cf587fa1cf1e0f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4a7a35b3dc00ec216c4ef53f0cf587fa1cf1e0f7"}]},"branch":"refs/heads/master"},"8431ed7405a585d4e7b5091a0c5b240888747175":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-08-27 22:38:09.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/97/805297/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/97/805297/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/97/805297/7"}}},"commit":{"parents":[{"commit":"4b5517a26878c709d4187253937aab54c95e2e33","subject":"Simplify test_pointer_device_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4b5517a26878c709d4187253937aab54c95e2e33"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-19 19:40:32.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-27 22:37:00.000000000","tz":-240},"subject":"Simplify test_rx_tx_queue_size","message":"Simplify test_rx_tx_queue_size\n\nDepend on tempest to create the network and subnet, and inline the\nserver creation to use the defaults and to remove the resource_setup()\n\nChange-Id: Ibc8a624153523aa14e073c46df2a5c793c5c230f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/8431ed7405a585d4e7b5091a0c5b240888747175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/8431ed7405a585d4e7b5091a0c5b240888747175"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
