)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~805453","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~I4d95a0a404e938a6e603548f83cfe3d734f5c613","project":"openstack/whitebox-tempest-plugin","branch":"master","topic":"interface_change_rx_queue","hashtags":[],"change_id":"I4d95a0a404e938a6e603548f83cfe3d734f5c613","subject":"Clean up .zuul.yaml","status":"MERGED","created":"2021-08-20 20:20:11.000000000","updated":"2021-08-28 23:17:44.000000000","submitted":"2021-08-28 23:17:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805453-interface_change_rx_queue","meta_rev_id":"a8fca5b96dacce58e62c0657d8797ef1c4ed0269","_number":805453,"virtual_id_number":805453,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-28 23:17:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":2,"date":"2021-08-28 23:17:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":1,"date":"2021-08-27 22:39:11.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-20 22:20:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 16:13:32.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"state":"REVIEWER"}],"messages":[{"id":"77f12b6b3da2e4f0761b8bd127982ede4cb56830","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-20 20:20:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af885b41395eb796fadee77f6f9afdc79ce74e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 22:20:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ba90b6824dd438a8f30b225a720d19d : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06424f8b59334da6a57cc315539345eb : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29e1460a94414a2f9a071eb3a793ef99 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/355ac6ab021d4860a62c8bfb31d5b692 : SUCCESS in 4m 14s (non-voting)\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/60903d7edbd74155be214da6a0d49482 : FAILURE in 1h 27m 54s","accounts_in_message":[],"_revision_number":1},{"id":"18e32704dfa8ede19bab5d277a528f16ad2ce1cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-20 22:29:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a366dfb674be8b0b355b85f0235e61d96244fb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 00:27:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5291c498d7914792b0f90d2aa282cfc2 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83203b18265c4727b48cf3d2d173f9ce : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/389d66f98d8b4c86abdb8e88bcfc161b : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a3e9559059a4f3086efc147fd220a95 : SUCCESS in 4m 22s (non-voting)\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/94963e8fa40c4978bbd99bfc2c45ce1f : SUCCESS in 1h 46m 39s","accounts_in_message":[],"_revision_number":2},{"id":"1107091aa058c81b292a30b4f455385d6ed8a212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-25 01:43:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7a20e9386b22757666d66897247ee1e0968103e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 03:33:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7c600478a64cb983fd18b302f0e4f7 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60acd7fbc23549bc896fa9ac8d929a81 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c0f96387e9342e99fa93ce4243506a9 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8be1b694da354062848fd787d4044259 : SUCCESS in 4m 10s (non-voting)\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/c3050e109ca840c7980e08a332e3ce87 : FAILURE in 1h 44m 34s","accounts_in_message":[],"_revision_number":3},{"id":"83ebfe6fb63fe52b5ffa08abd527a81d07e85564","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-25 16:51:43.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"734029a1b19bc53b8e1aeeda6e8e6326a71939e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 18:38:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34b0ed8a743b4c82ade19dd7db8a46e8 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e4913aed21c4c84809d6e006d32ad5b : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d71020fb384d4a51af4a51c49f25b296 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06a3ff86185a4cd7a4c6e0aec2440ae0 : SUCCESS in 7m 06s (non-voting)\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/e1d480ebdd6d401f9094dba7d0c5ed10 : SUCCESS in 1h 42m 17s","accounts_in_message":[],"_revision_number":4},{"id":"3b0166408086fd2f896bf6a861cfc163638c6b46","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2021-08-26 16:13:32.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7c40797586138335a5ae592e4ad8f3a5e82bd44f","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2021-08-27 15:08:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6aa683f7c077906cf7cccf6d3c203496d66a0ca9","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-27 15:34:13.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cdfe67422fe37e50588f2ac1c461e23ebabe24a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 15:35:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1ccdbcf6b9fd8f2c85ef58fe6f330f95d1a71e27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 17:43:16.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"e8eb80543c9e79ae60ebeda3b9f121ce188d2866","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-27 19:52:22.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"497a4e6f1ebe13812b91140f0f08b98279a75efe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 21:30:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a5793765e6488d91b439b67641e7ce : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/799cbe2f954a4644b928f9f618049333 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05043a30713742a1b0890ff8b6f5dac3 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a50f90d369f4c588cd071e9f1739f60 : SUCCESS in 5m 41s (non-voting)\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/aa3a80ee2d974b43a3de6abc9215710f : SUCCESS in 1h 35m 50s","accounts_in_message":[],"_revision_number":4},{"id":"1af4b8b3494de69d984da9d943991c8553991f2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 22:02:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bc2d8b923652114c3cb90512da3f48c460ecd7a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-27 22:38:09.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"66eb6be620c566ae08154ae71fc71789b304e9e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 22:38:28.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"30eb14558b7fa2f5ea55ade6d8f9c43ad0baecd0","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-27 22:39:11.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f3c4784d47dcb0fc4f5682d4c3beba51b8a51a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 00:35:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afc66895a6af436fb050ba519bc6311a : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f34b05db687247eca0a825ecb590db01 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3b814429e424e72ab267a425d143e57 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2a3879745c14ca4920ac9ea8629ca8b : SUCCESS in 4m 20s (non-voting)\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/d1c6ae5e4332411583f8741de8b13de0 : SUCCESS in 1h 52m 52s","accounts_in_message":[],"_revision_number":5},{"id":"0656ade44a29639a0067b79cd291b0438b6a29bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 21:39:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b2bc678fa3c38185ed4d9748316bbd37c8689284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 23:17:44.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07efe3b6979a466e9ef4de6291255ced : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6605c86ccf2f4d1e81a0046194266373 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c59e9cf6ee9d451fa788a89ae5f14036 : SUCCESS in 3m 35s\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/97c02075d8d14be785120fa9344f860d : SUCCESS in 1h 34m 29s","accounts_in_message":[],"_revision_number":5},{"id":"adf2c2d689e5ef84ab2cf811f01c50654f35bb9d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 23:17:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a8097361dd169a874539ca3249677723336a33b7","revisions":{"7b1c6bcd7982bfd65457dc5160d0291caff66bbb":{"kind":"REWORK","_number":1,"created":"2021-08-20 20:20:11.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/53/805453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/53/805453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/1"}}},"commit":{"parents":[{"commit":"ae8e5396d96d0a99ac918415d24bf44c1a5f4a82","subject":"Remove resize_server() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ae8e5396d96d0a99ac918415d24bf44c1a5f4a82"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-20 14:06:06.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-20 14:21:43.000000000","tz":-240},"subject":"Clean up .zuul.yaml","message":"Clean up .zuul.yaml\n\nRemove the multi-numa job definitions and nodeset, as we have nowhere\nto run them. Fix the indentation. Rename the jobs to\nwhitebox-devstack-multinode to stick to the\n\u003cproject\u003e-\u003cdeployment\u003e-\u003cenvionrment\u003e-\u003cfeature\u003e template, wich\n\u003cfeature\u003e becoming necessary in future patches.\n\nChange-Id: I4d95a0a404e938a6e603548f83cfe3d734f5c613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/7b1c6bcd7982bfd65457dc5160d0291caff66bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/7b1c6bcd7982bfd65457dc5160d0291caff66bbb"}]},"branch":"refs/heads/master"},"e003902150713df96f0e244cb021f158e9a8588d":{"kind":"REWORK","_number":2,"created":"2021-08-20 22:29:59.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/53/805453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/53/805453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/2"}}},"commit":{"parents":[{"commit":"ae8e5396d96d0a99ac918415d24bf44c1a5f4a82","subject":"Remove resize_server() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ae8e5396d96d0a99ac918415d24bf44c1a5f4a82"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-20 14:06:06.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-20 22:29:49.000000000","tz":-240},"subject":"Clean up .zuul.yaml","message":"Clean up .zuul.yaml\n\nRemove the multi-numa job definitions and nodeset, as we have nowhere\nto run them. Fix the indentation. Rename the jobs to\nwhitebox-devstack-multinode to stick to the\n\u003cproject\u003e-\u003cdeployment\u003e-\u003cenvionrment\u003e-\u003cfeature\u003e template, wich\n\u003cfeature\u003e becoming necessary in future patches.\n\nChange-Id: I4d95a0a404e938a6e603548f83cfe3d734f5c613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/e003902150713df96f0e244cb021f158e9a8588d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/e003902150713df96f0e244cb021f158e9a8588d"}]},"branch":"refs/heads/master"},"8f466bc0fb73251be9473bcb60a53e19686c3ef6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-08-25 01:43:45.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/53/805453/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/53/805453/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/3"}}},"commit":{"parents":[{"commit":"06d897cc222861c6f640df573304d32393037d5c","subject":"Use new config_all_computes() helper in file backed memory tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/06d897cc222861c6f640df573304d32393037d5c"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-20 14:06:06.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-25 01:42:45.000000000","tz":-240},"subject":"Clean up .zuul.yaml","message":"Clean up .zuul.yaml\n\nRemove the multi-numa job definitions and nodeset, as we have nowhere\nto run them. Fix the indentation. Rename the jobs to\nwhitebox-devstack-multinode to stick to the\n\u003cproject\u003e-\u003cdeployment\u003e-\u003cenvionrment\u003e-\u003cfeature\u003e template, wich\n\u003cfeature\u003e becoming necessary in future patches.\n\nChange-Id: I4d95a0a404e938a6e603548f83cfe3d734f5c613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/8f466bc0fb73251be9473bcb60a53e19686c3ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/8f466bc0fb73251be9473bcb60a53e19686c3ef6"}]},"branch":"refs/heads/master"},"6ba56b8e1f7523164a2d649925878526a26204ca":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-08-25 16:51:43.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/53/805453/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/53/805453/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/4"}}},"commit":{"parents":[{"commit":"87010eaa3711cc827d1e5be727c2a396857f7c89","subject":"Use new config_all_computes() helper in file backed memory tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/87010eaa3711cc827d1e5be727c2a396857f7c89"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-20 14:06:06.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-25 16:46:36.000000000","tz":-240},"subject":"Clean up .zuul.yaml","message":"Clean up .zuul.yaml\n\nRemove the multi-numa job definitions and nodeset, as we have nowhere\nto run them. Fix the indentation. Rename the jobs to\nwhitebox-devstack-multinode to stick to the\n\u003cproject\u003e-\u003cdeployment\u003e-\u003cenvionrment\u003e-\u003cfeature\u003e template, wich\n\u003cfeature\u003e becoming necessary in future patches.\n\nChange-Id: I4d95a0a404e938a6e603548f83cfe3d734f5c613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6ba56b8e1f7523164a2d649925878526a26204ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6ba56b8e1f7523164a2d649925878526a26204ca"}]},"branch":"refs/heads/master"},"a8097361dd169a874539ca3249677723336a33b7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-27 22:38:09.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/53/805453/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/53/805453/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/53/805453/5"}}},"commit":{"parents":[{"commit":"1532c496e745d355b8a2cc20d2f0040ed6eae4ad","subject":"Use new config_all_computes() helper in file backed memory tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/1532c496e745d355b8a2cc20d2f0040ed6eae4ad"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-20 14:06:06.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-27 22:37:00.000000000","tz":-240},"subject":"Clean up .zuul.yaml","message":"Clean up .zuul.yaml\n\nRemove the multi-numa job definitions and nodeset, as we have nowhere\nto run them. Fix the indentation. Rename the jobs to\nwhitebox-devstack-multinode to stick to the\n\u003cproject\u003e-\u003cdeployment\u003e-\u003cenvionrment\u003e-\u003cfeature\u003e template, wich\n\u003cfeature\u003e becoming necessary in future patches.\n\nChange-Id: I4d95a0a404e938a6e603548f83cfe3d734f5c613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a8097361dd169a874539ca3249677723336a33b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a8097361dd169a874539ca3249677723336a33b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
