)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~813058","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~Ib8023e7c7ab7b39233ab1245c0b39347cc2c4d71","project":"openstack/whitebox-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ib8023e7c7ab7b39233ab1245c0b39347cc2c4d71","subject":"DNM: Testing nested-virt-centos-8-stream","status":"ABANDONED","created":"2021-10-07 17:05:22.000000000","updated":"2021-11-09 10:53:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97241c9ce77843ac30950da3b7d19cdb24bb0fe0","_number":813058,"virtual_id_number":813058,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-07 18:45:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 18:45:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"141ffcd74d3c1ed3e877af8e9151de53921de429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-10-07 17:05:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c1d2cf45aee732867e11973b4d2bb3918202134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 18:45:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6fc478f4fc4d8a810c6ae9e9d5a5cd : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e61fe5b007944107a80f7ec71705b228 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37b36e90d6c540238d9b12dbf36c7e93 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e230c18dff144d6b6fa482faffd6403 : SUCCESS in 4m 18s (non-voting)\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/2bcd596ee6274231a43a8f80a7550e5a : FAILURE in 1h 35m 30s\n- whitebox-devstack-multinode-cpupinning https://zuul.opendev.org/t/openstack/build/1f834c38a17e480cabe3932cb66155cf : FAILURE in 40m 15s\n- whitebox-devstack-multinode-cpupinninglegacy https://zuul.opendev.org/t/openstack/build/6805227ac4f04405b6f6e433d1a04b3c : FAILURE in 40m 21s\n- whitebox-devstack-ceph-compute-local-ephemeral https://zuul.opendev.org/t/openstack/build/874a624b993944c1bf2aedcb9cbed7c9 : FAILURE in 47m 33s","accounts_in_message":[],"_revision_number":1},{"id":"97241c9ce77843ac30950da3b7d19cdb24bb0fe0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-09 10:53:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2757b1c0037e3480ff19557a723d0b9c413fac40","revisions":{"2757b1c0037e3480ff19557a723d0b9c413fac40":{"kind":"REWORK","_number":1,"created":"2021-10-07 17:05:22.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/58/813058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/58/813058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/813058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/813058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/813058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/58/813058/1"}}},"commit":{"parents":[{"commit":"d9492a943ac871dedc8bdafec4a3875077d43d71","subject":"Add Rbd direct download test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d9492a943ac871dedc8bdafec4a3875077d43d71"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-10-07 17:05:04.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-10-07 17:05:04.000000000","tz":60},"subject":"DNM: Testing nested-virt-centos-8-stream","message":"DNM: Testing nested-virt-centos-8-stream\n\nChange-Id: Ib8023e7c7ab7b39233ab1245c0b39347cc2c4d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2757b1c0037e3480ff19557a723d0b9c413fac40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2757b1c0037e3480ff19557a723d0b9c413fac40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
