)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~839117","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~I2a8c59ac4120bc618995bee51e8f43dbd073ee70","project":"openstack/whitebox-tempest-plugin","branch":"master","topic":"switch-to-zed-testing-runtime","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2022-05-04 17:10:12.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-05-04 15:56:43.000000000","reason":"Martin Kopec replied on the change"}},"hashtags":[],"change_id":"I2a8c59ac4120bc618995bee51e8f43dbd073ee70","subject":"Update to zed testing runtime","status":"MERGED","created":"2022-04-23 14:04:07.000000000","updated":"2022-05-04 17:10:12.000000000","submitted":"2022-05-04 17:10:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"839117-switch-to-zed-testing-runtime","meta_rev_id":"5d94ae43b22db26f7651a1c02c4a846fa98e01ed","_number":839117,"virtual_id_number":839117,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-04 17:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2022-05-04 15:56:43.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2022-05-04 15:56:43.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-23 15:15:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 14:33:39.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2022-05-04 15:56:43.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"f763d73a91dc3daa4bf039284255d3f1d3c528c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2022-04-23 14:04:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"271ab6eebba34d5cd0277cf050ae81105a603dcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-23 15:15:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1db98bc306104d688addf90f6779e342 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0d73a3df2cd4ab4a3f0b583653c593e : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2e7074bef4245f09f53ed61917dcbe7 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db487d1a15434dc5bc81a2ed334f9e5d : SUCCESS in 4m 07s (non-voting)\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/d3f191b726e043a6b0e55d6d2cba4274 : SUCCESS in 1h 09m 03s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/89b408ee5d624bae95375a7ad604958e : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":1},{"id":"14c499b3cc38dddddc2c0afdbd8829ee4f158d33","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-05-04 14:33:39.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"605ed442aee5b962e375e82516b0b23eafa48617","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-05-04 14:34:03.000000000","message":"Topic set to switch-to-zed-testing-runtime","accounts_in_message":[],"_revision_number":1},{"id":"3265720c562cb84a5265d3aadfa9150785bd7903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2022-05-04 14:44:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"966885a9ba6eadbe2f5a28932d0590b03d631827","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2022-05-04 14:45:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c630a30fb1c49793927f9576630dcc382b4e49d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 15:54:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86cd8fdc82bf4c7f97d1132e55c46b0a : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfc5334b041548f19daf83c93e96df18 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8963032e88b84fbead6c5ac2d25fe178 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac78e21861734eaf91756075080264c5 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c35e16f5b729454da148ca2f4149a8f0 : SUCCESS in 4m 14s\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/a83426a288e944c396ecf64371cddc5e : SUCCESS in 1h 07m 51s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/9db6788c148c4f2c887a082fec2e78d7 : SUCCESS in 46m 12s","accounts_in_message":[],"_revision_number":2},{"id":"d74a02e2edfc9a12e92e5ae36a2939c4abc868e9","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-05-04 15:56:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"094ef29b2d281297168525f4ddeba6bfb8f4ae3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 15:57:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8d76bcc38073fdea3799fd73605875220d865ef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 17:10:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80821bce8d2747beb3a94a90bff2f4a3 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f50695bdc4ea4119b3896e4941ad9e41 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dca2b16ce11f42b3a1d583ffc372e046 : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b539ee884ffc46aea1360b4245a4af78 : SUCCESS in 3m 31s\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/2e7b769e59b04a8eb78040099c35d902 : SUCCESS in 1h 10m 44s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/02916cf7049b416180e6e07d8108a45a : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":2},{"id":"5d94ae43b22db26f7651a1c02c4a846fa98e01ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 17:10:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8706a017f2b42d3b3534ef886afde8607a616f12","revisions":{"f40571d717636b92e5fd30a6ede44e84218094fb":{"kind":"REWORK","_number":1,"created":"2022-04-23 14:04:07.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/17/839117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/17/839117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/839117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/839117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/839117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/839117/1"}}},"commit":{"parents":[{"commit":"d8dde5e6d1dfcbd8228e514fb12c32f34ed6a0b7","subject":"Check for allocated or claimed PCI status in DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d8dde5e6d1dfcbd8228e514fb12c32f34ed6a0b7"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2022-04-23 14:03:56.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2022-04-23 14:03:56.000000000","tz":-240},"subject":"Update to zed testing runtime","message":"Update to zed testing runtime\n\nChange-Id: I2a8c59ac4120bc618995bee51e8f43dbd073ee70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/f40571d717636b92e5fd30a6ede44e84218094fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/f40571d717636b92e5fd30a6ede44e84218094fb"}]},"branch":"refs/heads/master"},"8706a017f2b42d3b3534ef886afde8607a616f12":{"kind":"REWORK","_number":2,"created":"2022-05-04 14:44:55.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/17/839117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/17/839117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/839117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/839117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/839117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/839117/2"}}},"commit":{"parents":[{"commit":"d8dde5e6d1dfcbd8228e514fb12c32f34ed6a0b7","subject":"Check for allocated or claimed PCI status in DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d8dde5e6d1dfcbd8228e514fb12c32f34ed6a0b7"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2022-04-23 14:03:56.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2022-05-04 14:44:12.000000000","tz":-240},"subject":"Update to zed testing runtime","message":"Update to zed testing runtime\n\nBecause whitebox also runs in other envs that still have Python 3.6,\nwhich the zed runtime removes, explicitly add it with\nopenstack-python36-jobs.\n\nChange-Id: I2a8c59ac4120bc618995bee51e8f43dbd073ee70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/8706a017f2b42d3b3534ef886afde8607a616f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/8706a017f2b42d3b3534ef886afde8607a616f12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
