)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~909225","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~I9d8f3c4148949bfdd9fc956e271c18c4d76183b6","project":"openstack/whitebox-tempest-plugin","branch":"master","topic":"windows_enlightenment_hyperv","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2024-02-19 21:04:52.000000000","reason":"Change was marked work in progress"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-02-19 21:04:52.000000000","reason":"Change was marked work in progress"},"31033":{"account":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"last_update":"2024-04-27 02:16:13.000000000","reason":"Change was submitted"},"29455":{"account":{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"},"last_update":"2024-02-19 21:04:52.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I9d8f3c4148949bfdd9fc956e271c18c4d76183b6","subject":"Add HyperV default enlightment tests","status":"MERGED","created":"2024-02-16 16:53:53.000000000","updated":"2024-04-27 02:16:13.000000000","submitted":"2024-04-27 02:16:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909225-windows_enlightenment_hyperv","meta_rev_id":"d03fbfa3272d6eebd5948412bcbb07f2f4cd328b","_number":909225,"virtual_id_number":909225,"owner":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-27 02:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":2,"date":"2024-04-27 01:01:34.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":1,"date":"2024-04-27 01:01:34.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"}],"reviewers":{"REVIEWER":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-16 18:03:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-16 19:28:09.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"},"state":"REVIEWER"},{"updated":"2024-02-16 19:28:09.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-02-16 19:28:09.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"}],"messages":[{"id":"276fc26e17f194bccc27188e7d25e8a5d180d347","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-02-16 16:53:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e3a6dfd62c583bf38f4f8924bbbeca0cbc467bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 18:03:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73e2cf1d1f0642a5b6fe70d3f111d1a6\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/23afbeb1aba4468999ece08be04c9293 : FAILURE in 1h 07m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c33f5369763d4363963ee1c88c5e17b8 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"96c51f91c5401fc298f373410a9dc880ff4d5ce0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-02-16 18:12:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2275d9cafc1e3cd7a678eace135e94ae34dbc589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:19:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5e0d2c9f5df4f6fb44b858752213287\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/587eb4ebda704552abdbd76e97ed96de : SUCCESS in 1h 06m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ec241a1fe14642abb18d580a62bc43 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"03e7ed165b9370d429b6d9744b88d3e15b20348a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-02-16 19:27:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"187b43eb8eaf75caf8fe7c6f53fd36d4fc233896","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-02-16 19:28:09.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"a93f6019be9576d282c02b0935d2d455d5e910f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 20:35:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de15d224f7f047498b481f074bce8ca0\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/f82e8dd42e684f918aa0c068bf69a977 : SUCCESS in 1h 06m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f50f12644deb4c6f92afdc7f60228af0 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":3},{"id":"adb2589fd1a9a14bb5eef6451910140b6d76359a","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-19 20:37:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bc35a25072587e6b9c099bef8a69645b13424662","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-02-19 20:43:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b1513323aac290e8d13f57fbe3557c30010c9304","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-19 20:51:35.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"432c2c773a8eab2b3e8f3d32eb14d8f3008c1242","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-02-19 21:01:39.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"03f7d5d8dee8c34a821079361a8399c78a198df8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-02-19 21:04:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0630265e142f6b0cbb057305f6a2cd499acf27ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 22:14:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41727b609be444a6ba45394e0fcdfcb6\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/81519508a46f4fbd84a8e6a3da6cbaa9 : FAILURE in 1h 09m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50b03c3f1341405784581719a88ed807 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":4},{"id":"fbe139166b8e905e12594bebc8c6f44874e0344c","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-20 18:20:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"128b05734eff0f51ab3cbb321334ea15ad3a780e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 19:39:05.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/260654e19df14628aa86aa7314911785\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/56eb65941ce341c79f97530b4a603f0c : FAILURE in 1h 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/092959183b1d4b05868e511b67396e8e : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":4},{"id":"10ce921293b20ded7e7fdd8f6747a7666cb99df4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-21 18:15:33.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d80e646426794698548fdbfb7f8a0397ffb4f358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 19:44:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4025c8b3680649c2a632129690f65b73\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/10d5ac21d7ac4a3f831b906e6727112b : FAILURE in 1h 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bef396aaf90d489d8ea1e276baf74672 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":5},{"id":"d2f9a01ec95f2410fa8c938552926f9e8542fba1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-21 20:20:09.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ac361a4c87cbf6cd508168384fdae4ff49af8a88","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-02-21 20:27:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bd0779cde3fed6d8eb9d8a13e749b054282afa7d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-21 21:05:39.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e844d314d737b51ae94e9aed9117ab6ab2858ce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 21:07:56.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/909785 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"c7030f866f83c0c4f3d092b5696f1312fd1c03bb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-21 21:08:23.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent tree, and commit message as Patch Set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"cb4eb0180d25531509e34d3429d5b3aec9c5e4a2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-02-21 21:09:16.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"988dd6184698eaac5b81cf061de8c03cc7e8ff6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 22:27:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e0f3047ff5645d8a415d106e926f99d\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/eef4da1d4e9442538e104a8a7d412fd2 : FAILURE in 1h 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2f53ba347c94e4baa8a019ae235b28a : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":10},{"id":"3b44ae61787248510ab52d5c4c5f49ea64ac2ac3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 01:02:22.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"f346fb8aff07ff3f3f6849ef4e94e64dc3f0344b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 01:22:42.000000000","message":"Uploaded patch set 12: New patch set was added with same tree, parent tree, and commit message as Patch Set 11.","accounts_in_message":[],"_revision_number":12},{"id":"430e103e51d903b0c45da17e667bbad8ec3eaddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 01:22:51.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/909785 is needed.","accounts_in_message":[],"_revision_number":11},{"id":"f2ca2cbb7fea325861b567fb9385fc9fd36f3114","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 01:24:19.000000000","message":"Uploaded patch set 13: New patch set was added with same tree, parent tree, and commit message as Patch Set 12.","accounts_in_message":[],"_revision_number":13},{"id":"e681caf835ca0676f6303f5d9acadf56c8968b6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 01:24:32.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/909785 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"79de2f82427dfa506fb3fd7a73b4d8c4409610d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 03:07:28.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1817b2fca4804e3280a890df11caefe4\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/7fbca96086ed41c2bd529cb7e8176d3c : FAILURE in 1h 34m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afd65f88a60c48cd99e2a97b4cc333ba : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":13},{"id":"d65796a71a1ce6d9a5929dd2e230314119877677","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 15:47:47.000000000","message":"Uploaded patch set 14: New patch set was added with same tree, parent tree, and commit message as Patch Set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"bba56f9ffe9556b4df4e4bbd5869f693adbdbc4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:16:32.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed0f3f3a384e42e98af6341e4c00a9a2\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/a645c89cb47140daaff707f24fed3c75 : FAILURE in 1h 25m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8992c95e154329b65b6e4d0957fdff : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":14},{"id":"25c2bfe42b45d79cc60d26078daf68a71ba50ef3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 18:33:26.000000000","message":"Uploaded patch set 15: New patch set was added with same tree, parent tree, and commit message as Patch Set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"0b135dfc297c0cbbb02eb6c8512422b152bb599b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 19:42:26.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2da4350530c94e41be9cc3de51918568\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/c6bc225b6f0f47d0b0901081dd229a8d : FAILURE in 1h 07m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d4caee18aa40e180a4f5e1e24fd31a : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":15},{"id":"0ec0af3702d4c8bf14dc4f6a7b140a4219c5c5a4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-23 01:39:49.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"dc5b3d1c959149bb90b7a167b668b18e85de17f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-23 01:46:32.000000000","message":"Uploaded patch set 17: New patch set was added with same tree, parent tree, and commit message as Patch Set 16.","accounts_in_message":[],"_revision_number":17},{"id":"767954c6ed37822607f8fb5a10319471bf50fee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 02:58:00.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/161db78996d94167af93075a36136a7b\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/44d499566b994d318eba9b9959ed7748 : SUCCESS in 1h 09m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b41afa4a4b464a439fe0ab29a1d909e2 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":17},{"id":"1de2589591724e0e2cb515675ec476399059aec2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-04-25 13:17:58.000000000","message":"Uploaded patch set 18: New patch set was added with same tree, parent tree, and commit message as Patch Set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"83a290b2c456ccb2d9c96dd1c32aadda2b869ae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 14:29:02.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b717d85117ef4f5c9f762862d0e9863d\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/0ab83865d15043128143e2804ca4ba10 : SUCCESS in 1h 09m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcb7ee5fa62b4284ae85ecc804f94971 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":18},{"id":"0cad67cd7873e5f40edccdd088e2204e58a0139d","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-04-26 19:42:00.000000000","message":"Patch Set 18:\n\n(2 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":18},{"id":"9d1a92924d47c19c59fdade13fa22cb245489ae6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-04-26 19:43:51.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"6b02a10497e96976106fa7d42e5e33bbfc65ee14","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-04-26 19:44:01.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"3d3c1c8b7c16a4b97b5f877acccdfec36a6c8cb7","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-04-26 19:44:17.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"b85c0c38b3ad7966353f52afd8ff8c87461a3f1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 20:51:16.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f48c6dd39e74d98b32fb76a09691d7d\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/94f5e5448dd14968ac94b59f2d664431 : SUCCESS in 1h 06m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/617a11702ffb4177a51bd5718a9f4412 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":19},{"id":"3b0cf2deb39879ff8203beb3a9477717e5226a65","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-04-27 01:01:34.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"b9f72a6eaeeda768578b05847b5b26e26a78e278","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-27 01:02:01.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"4476036392279917e5e8b8715414b9be152cca8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-27 02:16:13.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/822c22d4d65248d4b0403984981f981b\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/ea025f0cacbb47338cd5b80a98d753bc : SUCCESS in 1h 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ffb926a0ff4b21869c7505059b6a21 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":19},{"id":"d03fbfa3272d6eebd5948412bcbb07f2f4cd328b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-27 02:16:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"b40728768eb7d3f4f6eb8213830e4405e64dd58b","revisions":{"b0b5befc3dc62525fb91be881ec654f5bb17f5b9":{"kind":"REWORK","_number":1,"created":"2024-02-16 16:53:53.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/909225/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/1"}}},"commit":{"parents":[{"commit":"4b4ba3d8d914eb098362d421760b7bd215ee5a57","subject":"Merge \"Add test support for bochs display device\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4b4ba3d8d914eb098362d421760b7bd215ee5a57"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:21.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/b0b5befc3dc62525fb91be881ec654f5bb17f5b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/b0b5befc3dc62525fb91be881ec654f5bb17f5b9"}]},"branch":"refs/heads/master"},"1068a63746757af710dd148e91a7f0deb024fdda":{"kind":"REWORK","_number":2,"created":"2024-02-16 18:12:25.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/909225/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/2"}}},"commit":{"parents":[{"commit":"4b4ba3d8d914eb098362d421760b7bd215ee5a57","subject":"Merge \"Add test support for bochs display device\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4b4ba3d8d914eb098362d421760b7bd215ee5a57"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 18:12:22.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/1068a63746757af710dd148e91a7f0deb024fdda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/1068a63746757af710dd148e91a7f0deb024fdda"}]},"branch":"refs/heads/master"},"f2e79dd275cf134319cc8dc778a201f930c081d8":{"kind":"REWORK","_number":3,"created":"2024-02-16 19:27:43.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/909225/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/3"}}},"commit":{"parents":[{"commit":"4b4ba3d8d914eb098362d421760b7bd215ee5a57","subject":"Merge \"Add test support for bochs display device\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4b4ba3d8d914eb098362d421760b7bd215ee5a57"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 19:27:38.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/f2e79dd275cf134319cc8dc778a201f930c081d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/f2e79dd275cf134319cc8dc778a201f930c081d8"}]},"branch":"refs/heads/master"},"2cbb13cc574d68e51f0655d22d2d5d8896d889cd":{"kind":"REWORK","_number":4,"created":"2024-02-19 21:04:52.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/909225/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/4"}}},"commit":{"parents":[{"commit":"4b4ba3d8d914eb098362d421760b7bd215ee5a57","subject":"Merge \"Add test support for bochs display device\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4b4ba3d8d914eb098362d421760b7bd215ee5a57"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-19 21:02:41.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2cbb13cc574d68e51f0655d22d2d5d8896d889cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2cbb13cc574d68e51f0655d22d2d5d8896d889cd"}]},"branch":"refs/heads/master"},"2082fb5bcb77844ab726edb6dbf8e68eb6fbef5b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-21 18:15:33.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/5"}}},"commit":{"parents":[{"commit":"b2f4a45cce5f0996263e6c1afb161babf022f46a","subject":"POC: deploy with cpu_power_management\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/b2f4a45cce5f0996263e6c1afb161babf022f46a"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:15:30.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2082fb5bcb77844ab726edb6dbf8e68eb6fbef5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2082fb5bcb77844ab726edb6dbf8e68eb6fbef5b"}]},"branch":"refs/heads/master"},"df1eb5190b6a3dc638311848e65764185545833a":{"kind":"NO_CHANGE","_number":6,"created":"2024-02-21 20:20:09.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/6"}}},"commit":{"parents":[{"commit":"e6a1c532d8ca5c20ff8afb10f6453be6adb3840c","subject":"POC: deploy with cpu_power_management\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/e6a1c532d8ca5c20ff8afb10f6453be6adb3840c"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 20:20:04.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/df1eb5190b6a3dc638311848e65764185545833a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/df1eb5190b6a3dc638311848e65764185545833a"}]},"branch":"refs/heads/master"},"33fe5b764a4fab703b3dc8b8e66a743075243733":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":7,"created":"2024-02-21 20:27:43.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/909225/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/7"}}},"commit":{"parents":[{"commit":"4b4ba3d8d914eb098362d421760b7bd215ee5a57","subject":"Merge \"Add test support for bochs display device\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4b4ba3d8d914eb098362d421760b7bd215ee5a57"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-21 20:27:24.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/877773\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/33fe5b764a4fab703b3dc8b8e66a743075243733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/33fe5b764a4fab703b3dc8b8e66a743075243733"}]},"branch":"refs/heads/master"},"ad4bf55f3dc841a9b71b175609bd62464a9defd0":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":8,"created":"2024-02-21 21:05:39.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/8"}}},"commit":{"parents":[{"commit":"515a0a22186d1b9f0508d655f90bbb8c017bb12c","subject":"POC: deploy with cpu_power_management\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/515a0a22186d1b9f0508d655f90bbb8c017bb12c"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 21:05:33.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ad4bf55f3dc841a9b71b175609bd62464a9defd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ad4bf55f3dc841a9b71b175609bd62464a9defd0"}]},"branch":"refs/heads/master"},"35c5adf50ac8909a8a35f2b61ae2f17b694c18d2":{"kind":"NO_CHANGE","_number":9,"created":"2024-02-21 21:08:23.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/9"}}},"commit":{"parents":[{"commit":"6fb3aa96fa6fddd75fe49b3c7c06217d93169160","subject":"POC: deploy with cpu_power_management\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6fb3aa96fa6fddd75fe49b3c7c06217d93169160"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 21:08:19.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/35c5adf50ac8909a8a35f2b61ae2f17b694c18d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/35c5adf50ac8909a8a35f2b61ae2f17b694c18d2"}]},"branch":"refs/heads/master"},"f6efa7ed720f2dc029adea6faf92c5d5647639f1":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":10,"created":"2024-02-21 21:09:16.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/909225/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/10"}}},"commit":{"parents":[{"commit":"4b4ba3d8d914eb098362d421760b7bd215ee5a57","subject":"Merge \"Add test support for bochs display device\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/4b4ba3d8d914eb098362d421760b7bd215ee5a57"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-21 21:09:05.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/877773\nDepends-on: https://review.opendev.org/c/openstack/nova/+/909795\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/f6efa7ed720f2dc029adea6faf92c5d5647639f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/f6efa7ed720f2dc029adea6faf92c5d5647639f1"}]},"branch":"refs/heads/master"},"d8802db0bd8a0fa44883558c7042e69e158e4c48":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":11,"created":"2024-02-22 01:02:22.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/11"}}},"commit":{"parents":[{"commit":"ce292601969e2e26d063ecba412ecfe0975ef621","subject":"POC: deploy with cpu_power_management\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ce292601969e2e26d063ecba412ecfe0975ef621"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 01:02:19.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d8802db0bd8a0fa44883558c7042e69e158e4c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d8802db0bd8a0fa44883558c7042e69e158e4c48"}]},"branch":"refs/heads/master"},"42881613613ebc8e4e01075208a97cdd84fb7f3f":{"kind":"NO_CHANGE","_number":12,"created":"2024-02-22 01:22:42.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/12"}}},"commit":{"parents":[{"commit":"1469a2808c491853f24aada9cf6d943b5142491f","subject":"POC: deploy with cpu_power_management\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/1469a2808c491853f24aada9cf6d943b5142491f"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 01:22:38.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/42881613613ebc8e4e01075208a97cdd84fb7f3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/42881613613ebc8e4e01075208a97cdd84fb7f3f"}]},"branch":"refs/heads/master"},"89b18ea0a38f8d3d3acf398651c1246d565e7af4":{"kind":"NO_CHANGE","_number":13,"created":"2024-02-22 01:24:19.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/13"}}},"commit":{"parents":[{"commit":"d23a4da688368a613b886c89ed2499838434a20e","subject":"POC: deploy with cpu_power_management\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d23a4da688368a613b886c89ed2499838434a20e"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 01:24:12.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/89b18ea0a38f8d3d3acf398651c1246d565e7af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/89b18ea0a38f8d3d3acf398651c1246d565e7af4"}]},"branch":"refs/heads/master"},"1f26a75b181643f1323ab990f176d0be8c215b3c":{"kind":"NO_CHANGE","_number":14,"created":"2024-02-22 15:47:47.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/14"}}},"commit":{"parents":[{"commit":"499709d1e53debd98fb594c67780ebceea2f9cf9","subject":"POC: deploy with cpu_power_management\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/499709d1e53debd98fb594c67780ebceea2f9cf9"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 15:47:44.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/1f26a75b181643f1323ab990f176d0be8c215b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/1f26a75b181643f1323ab990f176d0be8c215b3c"}]},"branch":"refs/heads/master"},"6424d1d7f201c43b28d9552bc9d07533aff2ca92":{"kind":"NO_CHANGE","_number":15,"created":"2024-02-22 18:33:26.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/15"}}},"commit":{"parents":[{"commit":"2b8711fc929242eedb355c1c561a767a71573931","subject":"POC: deploy with cpu_power_management\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2b8711fc929242eedb355c1c561a767a71573931"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 18:33:21.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6424d1d7f201c43b28d9552bc9d07533aff2ca92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6424d1d7f201c43b28d9552bc9d07533aff2ca92"}]},"branch":"refs/heads/master"},"eb35bfdf62eebc9e113507200d89cb22587d2bf8":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-02-23 01:39:49.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/16"}}},"commit":{"parents":[{"commit":"36061dd7d3d0ccad48a0a0e9a08dd1e9aea244b9","subject":"Deploy with CPU power management enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/36061dd7d3d0ccad48a0a0e9a08dd1e9aea244b9"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-23 01:39:45.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/eb35bfdf62eebc9e113507200d89cb22587d2bf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/eb35bfdf62eebc9e113507200d89cb22587d2bf8"}]},"branch":"refs/heads/master"},"2190d471d10760f5dbce6095e355ac9a7304ba76":{"kind":"NO_CHANGE","_number":17,"created":"2024-02-23 01:46:32.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/17"}}},"commit":{"parents":[{"commit":"811bb1de12a63956c2852cf39ab48beec0bf200c","subject":"Deploy with CPU power management enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/811bb1de12a63956c2852cf39ab48beec0bf200c"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-23 01:46:04.000000000","tz":-300},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2190d471d10760f5dbce6095e355ac9a7304ba76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2190d471d10760f5dbce6095e355ac9a7304ba76"}]},"branch":"refs/heads/master"},"acb8043f6b5cff50af48d35b38c118ae79bbbff0":{"kind":"NO_CHANGE","_number":18,"created":"2024-04-25 13:17:58.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/18"}}},"commit":{"parents":[{"commit":"869fdccf31f33567f9980a7f22cd87eaafb24f05","subject":"Deploy with CPU power management enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/869fdccf31f33567f9980a7f22cd87eaafb24f05"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-04-25 13:17:48.000000000","tz":-240},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/acb8043f6b5cff50af48d35b38c118ae79bbbff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/acb8043f6b5cff50af48d35b38c118ae79bbbff0"}]},"branch":"refs/heads/master"},"b40728768eb7d3f4f6eb8213830e4405e64dd58b":{"kind":"REWORK","_number":19,"created":"2024-04-26 19:43:51.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/909225/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/909225/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/909225/19"}}},"commit":{"parents":[{"commit":"869fdccf31f33567f9980a7f22cd87eaafb24f05","subject":"Deploy with CPU power management enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/869fdccf31f33567f9980a7f22cd87eaafb24f05"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-02-16 16:45:18.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-04-26 19:43:47.000000000","tz":-240},"subject":"Add HyperV default enlightment tests","message":"Add HyperV default enlightment tests\n\nTests update of adding libvirt default enlightenments for HyperV when\ndeploying a guest that uses image os_type \u0027windows\u0027. [1] Test creates an\nimage and confirms the default attributes are present and in on state.\n\nIn addtion to cover regressions for [2], test that a guest with the\nenlightenments can live migrate as well.\n\n[1] https://review.opendev.org/c/openstack/nova/+/847641\n[2] https://review.opendev.org/c/openstack/nova/+/904183\n\nChange-Id: I9d8f3c4148949bfdd9fc956e271c18c4d76183b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/b40728768eb7d3f4f6eb8213830e4405e64dd58b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/b40728768eb7d3f4f6eb8213830e4405e64dd58b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
