)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~909785","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~Ia48289ee4f909a33bfe6dac19eccff558b448da9","project":"openstack/whitebox-tempest-plugin","branch":"master","topic":"windows_enlightenment_hyperv","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2024-04-26 19:01:04.000000000","reason":"Change was submitted"},"31033":{"account":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"last_update":"2024-04-26 19:01:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia48289ee4f909a33bfe6dac19eccff558b448da9","subject":"Deploy with CPU power management enabled","status":"MERGED","created":"2024-02-21 18:15:33.000000000","updated":"2024-04-26 19:01:04.000000000","submitted":"2024-04-26 19:01:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909785-windows_enlightenment_hyperv","meta_rev_id":"fdf75712ff746b3dafe60541fa04569328c5f1eb","_number":909785,"virtual_id_number":909785,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-26 19:01:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"all":[{"value":2,"date":"2024-04-26 17:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"all":[{"value":1,"date":"2024-04-26 17:55:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-21 19:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 13:08:00.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2024-04-25 15:49:44.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"state":"REVIEWER"}],"messages":[{"id":"37b35295e3d35e81c3b58209f075d910432478d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-21 18:15:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dacbcc4d2be8890c55b1a13e9fb48d85f4104aff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 19:52:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/601f23bfcfd24516b936536822fb61a0\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/532b30767df54147a01508e7616930f5 : FAILURE in 1h 05m 36s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/8a4f04716ba44ecc83a59c87d05c39d5 : SUCCESS in 48m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65d657452c5148a7ac114a508d5a3d4a : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4a6a6baaafe299f9a2141d68763d5c277909ea48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-21 20:20:09.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7eb8b32cfb1de9a907fdcee45830bb1411937c1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-21 21:05:39.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5e00be76c1765ae9fa9ce825f7bbd6cf241225f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 21:07:56.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/909795 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"f792ffa86cf9b4ef398b6243ada40578aefbedd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-21 21:08:23.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5fd154c2982e1d8494b605b4309b84681c81e869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 22:40:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d5be3cf605f4fd2949dfe6fa4dd0b92\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/e0976f88a5884f77a2251ee2e49647fb : FAILURE in 1h 23m 32s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/29d837c3a84b49cf81d9450ce02dff6a : SUCCESS in 34m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a20833aefe4594915aa8f898e14134 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":4},{"id":"7245221538f39511f66d4cad7f8cd20a9bcedc0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 01:02:22.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ff895423e9ef900c7bd11c3a426bb5a4d236499b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 01:22:42.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"1fcdc537d81b5c207c66f3782b24c61f539e6a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 01:22:50.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/909806 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"6105c77cd2a49db70ca525e8f619ff9cf77b808b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 01:24:19.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.","accounts_in_message":[],"_revision_number":7},{"id":"b39a9e3de45faf6d577f4f195b39655b39fc5960","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 01:24:31.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/909806 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"dbdb57af757a92927901dfcc4244e44fefec3539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 02:46:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca950ef003e8484bbede70486b62f107\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/20a6273dcbc34491aead4931eae0609e : FAILURE in 1h 21m 05s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/671b5b4885934e479a3fcbbc0b660649 : SUCCESS in 41m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a512209f1f054bfeae13a92448fdc1e8 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":7},{"id":"f1b3df8f50e31d247cc547b89689bf6cbcb5ae50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 02:59:54.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"c21be7b52839c31a7e5a87eae07b97301a85b2e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 04:22:52.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77c9f60402f84763af0a23684a4f4792\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/c79357bb6ecf46398b4b9267e9f8112f : FAILURE in 1h 20m 27s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/69ec281f408b4cf5b72e29507d224f03 : SUCCESS in 42m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea06c1898da94815a8856e952bf86d0c : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":8},{"id":"8015ba24b51fb88ef94d478dee1d66f4aef51782","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-02-22 13:08:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1c80c1d46169dbb83a9d6d97311a393a29fbf2e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 15:47:47.000000000","message":"Uploaded patch set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"5ea0669117982077607032b6d9cb871c21c4b1de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:11:37.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bd0a6a0309740e59d015b2eb242e45e\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/cbbe2fdbe8cc445da62cdf7fe6a4a000 : FAILURE in 1h 20m 34s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/f2958ce1e46c4ef690e728a567db2811 : SUCCESS in 42m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f0d5b843ea44bc9240da7179a16d09 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":9},{"id":"456b3d1aab94ddcb4c2414108b617ef61870d450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 18:33:26.000000000","message":"Uploaded patch set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"df31c3b0fc7a8fcfbb6a89abe4e785a4d1b0ba67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 19:40:12.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c759580ecfc4c7bb42f8f6dbc441f3a\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/0a3d3048b886469eba2889f000cb47e7 : FAILURE in 1h 05m 06s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/c5148aace6d94bad8c082b5c98c859fc : SUCCESS in 48m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1818d26b1623426eac8f054d31d6e90d : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":10},{"id":"f6ef00b9f09ae63381f031b67cb2cfab108859c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-23 01:39:49.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"9ac2ce71403432d0d976641598d43eae7506df79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-23 01:46:32.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"1f6e85240f285892d1c3f8de52c56a2b90c11c80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 02:54:52.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61f4950fd76a475e9567699f1cbc573e\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/f9896e538f7c40869a557dc61f97a394 : SUCCESS in 1h 06m 56s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/38b9d1ab03a54080bb243da024b41522 : SUCCESS in 54m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b573f0f0a34999815b6b87306010ab : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":12},{"id":"1d7380e3a0cf7bdbc1b8591a4cd72a452979fcb9","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-03-09 01:44:51.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"dccb15221a22b9929c0e1608867b019871077131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-09 01:45:20.000000000","message":"Patch Set 12: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 877773,8","accounts_in_message":[],"_revision_number":12},{"id":"64e7c3cf3000f739c00e1a16d9a74be64f4e40ef","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-03-12 14:19:09.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"78c020e8ed99aad9f1af3f1eb9ecd2b7761063ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 15:25:30.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3794dbf056f349f7bd367b94b369f1a2\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/7725876e405a413b87523a095ac89e86 : SUCCESS in 1h 04m 54s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/a89619ddac904873a315193fc8ca95e7 : SUCCESS in 47m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e3a22380e04e05b51044686a6083fe : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":12},{"id":"481b8e8fe8a2b544575f2f22db5aa5368008866f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-04-25 13:17:58.000000000","message":"Uploaded patch set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"fbb3271e6f62d1773f6f372b2588d4949d0a0d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:49:01.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df829f35e8954c608963eba23c74ea10\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/31c40964bf394cd9a95e21aa3a80eb9c : SUCCESS in 1h 06m 09s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/bc079da539c64349bd8d7dbadb58762d : SUCCESS in 1h 01m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ade6ded7ec85401fbdace0c3893f5ae1 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":13},{"id":"094149eff998afb8e738afc9185dadb9fbc4d1d5","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-04-26 17:55:22.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"5d22abb0476fde90c654f09f1e65a2d9a1827dfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 17:55:59.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"f77d33161051a2bfbab5ac9ce0b8e51fd0548706","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 19:01:04.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18f7504db7c04374a16fce5c166c5561\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/1ef1e9c0c1854b78b47e5eab5f49971e : SUCCESS in 1h 04m 31s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/b4bbbc288eac49ecbf7e62227f0a78bc : SUCCESS in 46m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/257d0dae4ce5428386f3777f314c5186 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":13},{"id":"fdf75712ff746b3dafe60541fa04569328c5f1eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 19:01:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"869fdccf31f33567f9980a7f22cd87eaafb24f05","revisions":{"b2f4a45cce5f0996263e6c1afb161babf022f46a":{"kind":"REWORK","_number":1,"created":"2024-02-21 18:15:33.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/1"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"subject":"POC: deploy with cpu_power_management\u003dTrue","message":"POC: deploy with cpu_power_management\u003dTrue\n\nIn the patch above this one, we\u0027re getting a mysterious failure in\npost_live_migraton, where the pci_tracker appears to not be\ninitialized:\n\nFile \"/opt/stack/nova/nova/compute/resource_tracker.py\", line 2089, in allocate_pci_devices_for_instance\nself.pci_tracker.allocate_instance(instance)\n\u0027NoneType\u0027 object has no attribute \u0027allocate_instance\u0027\n\nThere\u0027s nothing out of the ordinary in the logs preceeding this. The\nonly potentially suspicious thing is the following, long before:\n\nERROR nova.compute.manager nova.exception.Invalid: Invalid \u0027[compute] cpu_dedicated_set\u0027 config: one or more of the configured CPUs is not available. Available cpuset(s): [0, 4, 5, 6, 7], configured cpuset(s): [1, 2, 3]\n\nWe might be doing some wrong when we enable CPU power management and\nrestart nova-compute, leading to the resource tracker (and PCI\nmanager) getting broken.\n\nThis patch tries to run the CPU power management tests modified to\nassume a pre-configured deployment.\n\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/b2f4a45cce5f0996263e6c1afb161babf022f46a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/b2f4a45cce5f0996263e6c1afb161babf022f46a"}]},"branch":"refs/heads/master"},"e6a1c532d8ca5c20ff8afb10f6453be6adb3840c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-21 20:20:09.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/2"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 20:19:59.000000000","tz":-300},"subject":"POC: deploy with cpu_power_management\u003dTrue","message":"POC: deploy with cpu_power_management\u003dTrue\n\nIn the patch above this one, we\u0027re getting a mysterious failure in\npost_live_migraton, where the pci_tracker appears to not be\ninitialized:\n\nFile \"/opt/stack/nova/nova/compute/resource_tracker.py\", line 2089, in allocate_pci_devices_for_instance\nself.pci_tracker.allocate_instance(instance)\n\u0027NoneType\u0027 object has no attribute \u0027allocate_instance\u0027\n\nThere\u0027s nothing out of the ordinary in the logs preceeding this. The\nonly potentially suspicious thing is the following, long before:\n\nERROR nova.compute.manager nova.exception.Invalid: Invalid \u0027[compute] cpu_dedicated_set\u0027 config: one or more of the configured CPUs is not available. Available cpuset(s): [0, 4, 5, 6, 7], configured cpuset(s): [1, 2, 3]\n\nWe might be doing some wrong when we enable CPU power management and\nrestart nova-compute, leading to the resource tracker (and PCI\nmanager) getting broken.\n\nThis patch tries to run the CPU power management tests modified to\nassume a pre-configured deployment.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/877773\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/e6a1c532d8ca5c20ff8afb10f6453be6adb3840c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/e6a1c532d8ca5c20ff8afb10f6453be6adb3840c"}]},"branch":"refs/heads/master"},"515a0a22186d1b9f0508d655f90bbb8c017bb12c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-02-21 21:05:39.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/3"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 21:05:19.000000000","tz":-300},"subject":"POC: deploy with cpu_power_management\u003dTrue","message":"POC: deploy with cpu_power_management\u003dTrue\n\nIn the patch above this one, we\u0027re getting a mysterious failure in\npost_live_migraton, where the pci_tracker appears to not be\ninitialized:\n\nFile \"/opt/stack/nova/nova/compute/resource_tracker.py\", line 2089, in allocate_pci_devices_for_instance\nself.pci_tracker.allocate_instance(instance)\n\u0027NoneType\u0027 object has no attribute \u0027allocate_instance\u0027\n\nThere\u0027s nothing out of the ordinary in the logs preceeding this. The\nonly potentially suspicious thing is the following, long before:\n\nERROR nova.compute.manager nova.exception.Invalid: Invalid \u0027[compute] cpu_dedicated_set\u0027 config: one or more of the configured CPUs is not available. Available cpuset(s): [0, 4, 5, 6, 7], configured cpuset(s): [1, 2, 3]\n\nWe might be doing some wrong when we enable CPU power management and\nrestart nova-compute, leading to the resource tracker (and PCI\nmanager) getting broken.\n\nThis patch tries to run the CPU power management tests modified to\nassume a pre-configured deployment.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/877773\nDepends-on: https://review.opendev.org/c/openstack/nova/+/909795\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/515a0a22186d1b9f0508d655f90bbb8c017bb12c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/515a0a22186d1b9f0508d655f90bbb8c017bb12c"}]},"branch":"refs/heads/master"},"6fb3aa96fa6fddd75fe49b3c7c06217d93169160":{"kind":"NO_CHANGE","_number":4,"created":"2024-02-21 21:08:23.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/4"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 21:08:16.000000000","tz":-300},"subject":"POC: deploy with cpu_power_management\u003dTrue","message":"POC: deploy with cpu_power_management\u003dTrue\n\nIn the patch above this one, we\u0027re getting a mysterious failure in\npost_live_migraton, where the pci_tracker appears to not be\ninitialized:\n\nFile \"/opt/stack/nova/nova/compute/resource_tracker.py\", line 2089, in allocate_pci_devices_for_instance\nself.pci_tracker.allocate_instance(instance)\n\u0027NoneType\u0027 object has no attribute \u0027allocate_instance\u0027\n\nThere\u0027s nothing out of the ordinary in the logs preceeding this. The\nonly potentially suspicious thing is the following, long before:\n\nERROR nova.compute.manager nova.exception.Invalid: Invalid \u0027[compute] cpu_dedicated_set\u0027 config: one or more of the configured CPUs is not available. Available cpuset(s): [0, 4, 5, 6, 7], configured cpuset(s): [1, 2, 3]\n\nWe might be doing some wrong when we enable CPU power management and\nrestart nova-compute, leading to the resource tracker (and PCI\nmanager) getting broken.\n\nThis patch tries to run the CPU power management tests modified to\nassume a pre-configured deployment.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/877773\nDepends-on: https://review.opendev.org/c/openstack/nova/+/909795\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6fb3aa96fa6fddd75fe49b3c7c06217d93169160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6fb3aa96fa6fddd75fe49b3c7c06217d93169160"}]},"branch":"refs/heads/master"},"ce292601969e2e26d063ecba412ecfe0975ef621":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-02-22 01:02:22.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/5"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 01:02:12.000000000","tz":-300},"subject":"POC: deploy with cpu_power_management\u003dTrue","message":"POC: deploy with cpu_power_management\u003dTrue\n\nIn the patch above this one, we\u0027re getting a mysterious failure in\npost_live_migraton, where the pci_tracker appears to not be\ninitialized:\n\nFile \"/opt/stack/nova/nova/compute/resource_tracker.py\", line 2089, in allocate_pci_devices_for_instance\nself.pci_tracker.allocate_instance(instance)\n\u0027NoneType\u0027 object has no attribute \u0027allocate_instance\u0027\n\nThere\u0027s nothing out of the ordinary in the logs preceeding this. The\nonly potentially suspicious thing is the following, long before:\n\nERROR nova.compute.manager nova.exception.Invalid: Invalid \u0027[compute] cpu_dedicated_set\u0027 config: one or more of the configured CPUs is not available. Available cpuset(s): [0, 4, 5, 6, 7], configured cpuset(s): [1, 2, 3]\n\nWe might be doing some wrong when we enable CPU power management and\nrestart nova-compute, leading to the resource tracker (and PCI\nmanager) getting broken.\n\nThis patch tries to run the CPU power management tests modified to\nassume a pre-configured deployment.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/909806\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ce292601969e2e26d063ecba412ecfe0975ef621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ce292601969e2e26d063ecba412ecfe0975ef621"}]},"branch":"refs/heads/master"},"1469a2808c491853f24aada9cf6d943b5142491f":{"kind":"NO_CHANGE","_number":6,"created":"2024-02-22 01:22:42.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/6"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 01:22:37.000000000","tz":-300},"subject":"POC: deploy with cpu_power_management\u003dTrue","message":"POC: deploy with cpu_power_management\u003dTrue\n\nIn the patch above this one, we\u0027re getting a mysterious failure in\npost_live_migraton, where the pci_tracker appears to not be\ninitialized:\n\nFile \"/opt/stack/nova/nova/compute/resource_tracker.py\", line 2089, in allocate_pci_devices_for_instance\nself.pci_tracker.allocate_instance(instance)\n\u0027NoneType\u0027 object has no attribute \u0027allocate_instance\u0027\n\nThere\u0027s nothing out of the ordinary in the logs preceeding this. The\nonly potentially suspicious thing is the following, long before:\n\nERROR nova.compute.manager nova.exception.Invalid: Invalid \u0027[compute] cpu_dedicated_set\u0027 config: one or more of the configured CPUs is not available. Available cpuset(s): [0, 4, 5, 6, 7], configured cpuset(s): [1, 2, 3]\n\nWe might be doing some wrong when we enable CPU power management and\nrestart nova-compute, leading to the resource tracker (and PCI\nmanager) getting broken.\n\nThis patch tries to run the CPU power management tests modified to\nassume a pre-configured deployment.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/909806\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/1469a2808c491853f24aada9cf6d943b5142491f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/1469a2808c491853f24aada9cf6d943b5142491f"}]},"branch":"refs/heads/master"},"d23a4da688368a613b886c89ed2499838434a20e":{"kind":"NO_CHANGE","_number":7,"created":"2024-02-22 01:24:19.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/7"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 01:24:10.000000000","tz":-300},"subject":"POC: deploy with cpu_power_management\u003dTrue","message":"POC: deploy with cpu_power_management\u003dTrue\n\nIn the patch above this one, we\u0027re getting a mysterious failure in\npost_live_migraton, where the pci_tracker appears to not be\ninitialized:\n\nFile \"/opt/stack/nova/nova/compute/resource_tracker.py\", line 2089, in allocate_pci_devices_for_instance\nself.pci_tracker.allocate_instance(instance)\n\u0027NoneType\u0027 object has no attribute \u0027allocate_instance\u0027\n\nThere\u0027s nothing out of the ordinary in the logs preceeding this. The\nonly potentially suspicious thing is the following, long before:\n\nERROR nova.compute.manager nova.exception.Invalid: Invalid \u0027[compute] cpu_dedicated_set\u0027 config: one or more of the configured CPUs is not available. Available cpuset(s): [0, 4, 5, 6, 7], configured cpuset(s): [1, 2, 3]\n\nWe might be doing some wrong when we enable CPU power management and\nrestart nova-compute, leading to the resource tracker (and PCI\nmanager) getting broken.\n\nThis patch tries to run the CPU power management tests modified to\nassume a pre-configured deployment.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/909806\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d23a4da688368a613b886c89ed2499838434a20e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d23a4da688368a613b886c89ed2499838434a20e"}]},"branch":"refs/heads/master"},"cba4ea46db2cb4f3226024061727a7b039ca4de7":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2024-02-22 02:59:54.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/8"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 02:59:54.000000000","tz":0},"subject":"POC: deploy with cpu_power_management\u003dTrue","message":"POC: deploy with cpu_power_management\u003dTrue\n\nIn the patch above this one, we\u0027re getting a mysterious failure in\npost_live_migraton, where the pci_tracker appears to not be\ninitialized:\n\nFile \"/opt/stack/nova/nova/compute/resource_tracker.py\", line 2089, in allocate_pci_devices_for_instance\nself.pci_tracker.allocate_instance(instance)\n\u0027NoneType\u0027 object has no attribute \u0027allocate_instance\u0027\n\nThere\u0027s nothing out of the ordinary in the logs preceeding this. The\nonly potentially suspicious thing is the following, long before:\n\nERROR nova.compute.manager nova.exception.Invalid: Invalid \u0027[compute] cpu_dedicated_set\u0027 config: one or more of the configured CPUs is not available. Available cpuset(s): [0, 4, 5, 6, 7], configured cpuset(s): [1, 2, 3]\n\nWe might be doing some wrong when we enable CPU power management and\nrestart nova-compute, leading to the resource tracker (and PCI\nmanager) getting broken.\n\nThis patch tries to run the CPU power management tests modified to\nassume a pre-configured deployment.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/877773\nDepends-on: https://review.opendev.org/c/openstack/nova/+/909806\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/cba4ea46db2cb4f3226024061727a7b039ca4de7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/cba4ea46db2cb4f3226024061727a7b039ca4de7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"499709d1e53debd98fb594c67780ebceea2f9cf9":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-02-22 15:47:47.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/9"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 15:47:36.000000000","tz":-300},"subject":"POC: deploy with cpu_power_management\u003dTrue","message":"POC: deploy with cpu_power_management\u003dTrue\n\nIn the patch above this one, we\u0027re getting a mysterious failure in\npost_live_migraton, where the pci_tracker appears to not be\ninitialized:\n\nFile \"/opt/stack/nova/nova/compute/resource_tracker.py\", line 2089, in allocate_pci_devices_for_instance\nself.pci_tracker.allocate_instance(instance)\n\u0027NoneType\u0027 object has no attribute \u0027allocate_instance\u0027\n\nThere\u0027s nothing out of the ordinary in the logs preceeding this. The\nonly potentially suspicious thing is the following, long before:\n\nERROR nova.compute.manager nova.exception.Invalid: Invalid \u0027[compute] cpu_dedicated_set\u0027 config: one or more of the configured CPUs is not available. Available cpuset(s): [0, 4, 5, 6, 7], configured cpuset(s): [1, 2, 3]\n\nWe might be doing some wrong when we enable CPU power management and\nrestart nova-compute, leading to the resource tracker (and PCI\nmanager) getting broken.\n\nThis patch tries to run the CPU power management tests modified to\nassume a pre-configured deployment.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/909806\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/499709d1e53debd98fb594c67780ebceea2f9cf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/499709d1e53debd98fb594c67780ebceea2f9cf9"}]},"branch":"refs/heads/master"},"2b8711fc929242eedb355c1c561a767a71573931":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2024-02-22 18:33:26.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/10"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 18:33:03.000000000","tz":-300},"subject":"POC: deploy with cpu_power_management\u003dTrue","message":"POC: deploy with cpu_power_management\u003dTrue\n\nIn the patch above this one, we\u0027re getting a mysterious failure in\npost_live_migraton, where the pci_tracker appears to not be\ninitialized:\n\nFile \"/opt/stack/nova/nova/compute/resource_tracker.py\", line 2089, in allocate_pci_devices_for_instance\nself.pci_tracker.allocate_instance(instance)\n\u0027NoneType\u0027 object has no attribute \u0027allocate_instance\u0027\n\nThere\u0027s nothing out of the ordinary in the logs preceeding this. The\nonly potentially suspicious thing is the following, long before:\n\nERROR nova.compute.manager nova.exception.Invalid: Invalid \u0027[compute] cpu_dedicated_set\u0027 config: one or more of the configured CPUs is not available. Available cpuset(s): [0, 4, 5, 6, 7], configured cpuset(s): [1, 2, 3]\n\nWe might be doing some wrong when we enable CPU power management and\nrestart nova-compute, leading to the resource tracker (and PCI\nmanager) getting broken.\n\nThis patch tries to run the CPU power management tests modified to\nassume a pre-configured deployment.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/877773\nDepends-on: https://review.opendev.org/c/openstack/nova/+/909806\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2b8711fc929242eedb355c1c561a767a71573931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2b8711fc929242eedb355c1c561a767a71573931"}]},"branch":"refs/heads/master"},"36061dd7d3d0ccad48a0a0e9a08dd1e9aea244b9":{"kind":"REWORK","_number":11,"created":"2024-02-23 01:39:49.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/11"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-23 01:34:15.000000000","tz":-300},"subject":"Deploy with CPU power management enabled","message":"Deploy with CPU power management enabled\n\nTurns out CPU power management is broken with the `isolate` thread\nemualtor thread policy, and with live migration. This patch turns on\nCPU power management in our CI job, and modifies the test to assume\nthat it\u0027s turned on. It depends on all the Nova fixes needed to make\nit work.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/877773\nDepends-on: https://review.opendev.org/c/openstack/nova/+/909806\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/36061dd7d3d0ccad48a0a0e9a08dd1e9aea244b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/36061dd7d3d0ccad48a0a0e9a08dd1e9aea244b9"}]},"branch":"refs/heads/master"},"811bb1de12a63956c2852cf39ab48beec0bf200c":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2024-02-23 01:46:32.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/12"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-23 01:41:07.000000000","tz":-300},"subject":"Deploy with CPU power management enabled","message":"Deploy with CPU power management enabled\n\nTurns out CPU power management is broken with the `isolate` thread\nemualtor thread policy, and with live migration. This patch turns on\nCPU power management in our CI job, and modifies the test to assume\nthat it\u0027s turned on. It depends on all the Nova fixes needed to make\nit work.\n\nWe pick up the fix for live migration with cpu_shared_set. In the\npower management use case, it\u0027s necessary because it ensures we update\nthe host CPUs to pin the guest floating CPUs to. Without it, we could\nattempt to pin to a CPU that was shared on the source, but is\ndedicated and powerd off on the destination. This patch therefore\nclears the FIXME in that test.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/877773\nDepends-on: https://review.opendev.org/c/openstack/nova/+/909806\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/811bb1de12a63956c2852cf39ab48beec0bf200c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/811bb1de12a63956c2852cf39ab48beec0bf200c"}]},"branch":"refs/heads/master"},"869fdccf31f33567f9980a7f22cd87eaafb24f05":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2024-04-25 13:17:58.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/85/909785/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/85/909785/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/85/909785/13"}}},"commit":{"parents":[{"commit":"766ff042bd8528043947f16c84d13c9645d258d1","subject":"Add vIOMMU device tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/766ff042bd8528043947f16c84d13c9645d258d1"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-21 18:11:45.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-04-25 13:12:49.000000000","tz":-240},"subject":"Deploy with CPU power management enabled","message":"Deploy with CPU power management enabled\n\nUntil we fixed https://bugs.launchpad.net/nova/+bug/2056613 and\nhttps://bugs.launchpad.net/nova/+bug/2056612, libvirt CPU power\nmanagement was broken with the `isolate` thread emualtor thread\npolicy, and with live migration. This was noticed more or less by\naccident when the patch above this one triggered the right conditions\nfor our existing tests with incomplete coverage to break.\n\nTo prevent his from happening again, this patch turns on CPU power\nmanagement in our CI job, and modifies the test to assume that it\u0027s\nturned on.\n\nChange-Id: Ia48289ee4f909a33bfe6dac19eccff558b448da9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/869fdccf31f33567f9980a7f22cd87eaafb24f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/869fdccf31f33567f9980a7f22cd87eaafb24f05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
