)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~927525","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~I2cf0ded367e85fca1e49a2cab48adb23b1ed9807","project":"openstack/whitebox-tempest-plugin","branch":"master","topic":"update_rbd_log_query","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2024-10-08 00:14:25.000000000","reason":"\u003cGERRIT_ACCOUNT_8864\u003e replied on the change","reason_account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}},"35153":{"account":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"last_update":"2024-10-08 02:03:38.000000000","reason":"Change was submitted"},"34309":{"account":{"_account_id":34309,"name":"Sreelakshmi Menon Kovili","display_name":"Sree","email":"skovili@redhat.com","username":"sreekovili"},"last_update":"2024-10-08 02:03:38.000000000","reason":"Change was submitted"},"31033":{"account":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"last_update":"2024-10-08 02:03:38.000000000","reason":"Change was submitted"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2024-10-08 02:03:38.000000000","reason":"Change was submitted"},"29455":{"account":{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"},"last_update":"2024-10-08 02:03:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2cf0ded367e85fca1e49a2cab48adb23b1ed9807","subject":"Update accessing nova compute logs for RBD check","status":"MERGED","created":"2024-08-29 15:51:26.000000000","updated":"2024-10-08 02:03:38.000000000","submitted":"2024-10-08 02:03:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927525-update_rbd_log_query","meta_rev_id":"b79ff47185849278016273fa74002c1af215a480","_number":927525,"virtual_id_number":927525,"owner":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-08 02:03:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":2,"date":"2024-10-08 00:14:25.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":1,"date":"2024-10-08 00:14:25.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},{"_account_id":34309,"name":"Sreelakshmi Menon Kovili","display_name":"Sree","email":"skovili@redhat.com","username":"sreekovili"},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"}],"reviewers":{"REVIEWER":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"},{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"_account_id":34309,"name":"Sreelakshmi Menon Kovili","display_name":"Sree","email":"skovili@redhat.com","username":"sreekovili"},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-29 18:45:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-25 16:37:51.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2024-09-25 16:37:51.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"},"state":"REVIEWER"},{"updated":"2024-09-25 16:37:51.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"state":"REVIEWER"},{"updated":"2024-09-25 16:37:51.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":34309,"name":"Sreelakshmi Menon Kovili","display_name":"Sree","email":"skovili@redhat.com","username":"sreekovili"},"state":"REVIEWER"},{"updated":"2024-09-25 16:37:51.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"}],"messages":[{"id":"841f12b96710cc996b45ce925c6c0f355eaca099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-08-29 15:51:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"503a9a1d9a4e6950dc09f8cc7e46c7b2a59fe503","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-08-29 16:01:44.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4f1a73ad8121b2c486d49407d46d61a5891cd427","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-08-29 16:33:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9cda2bf0a8c39e0071c35404a9e32b33f38495ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 18:45:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aa07dc59fee48438e7d639b765676c9\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/e97e77e21442435792c95260fa9de28a : SUCCESS in 2h 03m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df88d27000af4ae3b77945a68c71ebfd : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"366a9a2c2447e630e097a95d19a1b1780e1196ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-09-24 16:50:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7a637fc56b3c84632107f60921f5a1599367b80a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 17:36:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cb6decb94fa4a29a63761174c1b7003\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/42f80d0bcfac497096ea0f4e3ad43450 : FAILURE in 44m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd70e42f091412ea5522077ac72bd95 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":3},{"id":"eea35636c8e308d1e10a2368d6b3239268981dc2","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-09-25 14:32:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4ac1221c529ec7d175433f877345114a7bb73b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 16:26:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/390ef7ab06a449eaa011647e9765ef46\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/438d265e5e0a46fe89ccc5b1d41af12f : SUCCESS in 1h 51m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d51dbca095954f2e8a33bf1805301e53 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":3},{"id":"9bdeec4827ae04292bb8d9a0c238479c41a48661","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-09-25 16:37:51.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"5112713c4eeefc667461475ac59bed77dd9ab4f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-09-25 18:37:02.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f9f7b301838689f5f367048747f55167aeea15bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 20:33:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66e82e8ab8ff42d8ab5b05ff48053432\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/e02977b03c8c42a2b17edfb81c4f762a : SUCCESS in 1h 55m 45s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/81d156a6f87741a1b9146357e9b30935 : FAILURE in 47m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9443be805654d85892b8d409316d1b0 : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":4},{"id":"b973782d06ed77af1e58fc2ff76e1d24807d567c","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-09-26 14:32:05.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2f93a2e7456102a67916892edf628b94be7887aa","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-09-26 15:15:54.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2e1409feca8106ae05f835c88be46669c9079f54","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-09-26 15:22:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"123b5bfa809083263f6709136c20f335580c651b","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-26 15:38:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"04b86c2cd65ba05643c393e206eb61770ef3468b","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-26 15:39:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"04a2507bb4d697a109bf7971652ed62686d68403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 17:57:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3a6e1b704e5452c9963f7ffeed2dab0\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/6c3e936beefe4c4d9f5c987433e78ebd : SUCCESS in 1h 56m 47s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/20f7bf4d5cb14db2a4d6f70535b6cec8 : FAILURE in 39m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff0f9809e72481796241a2d343b867e : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":5},{"id":"ce5a8980768272d8428742028e975908a24bb382","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-09-26 19:38:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f0513c5f134d5aede713f0ce32b483014aec4fe5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-09-30 16:25:18.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c321cebafe6ed1330b4cdd96479224696a68cc81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 18:32:12.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76cd460a62694acbb718bd0b55f1587a\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/c4c2982e77054460ac002aacffecd3f7 : SUCCESS in 1h 57m 50s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/9dd8db0c72cf4d54a42516f6cd36ecea : FAILURE in 52m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e32a466b7394b638d5435470e73e870 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":6},{"id":"2a33df97d32f175957ed792fb487918313a9c31d","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-09-30 22:53:47.000000000","message":"Patch Set 6:\n\n(3 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"89fabfc39f550adbba232b6962b514b77562020a","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-10-01 16:31:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dfc869341d1417b00d2db689a723f0dc06907da4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 18:42:05.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97ddbceeac3d4ae58018da9c110b494a\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/3b4dbdbef9374834aaa82182021b2f06 : SUCCESS in 2h 08m 51s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/e027d2b37c4348ad9e957f51df4b2c73 : FAILURE in 46m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37dbae2df4234ae4b0dd4cf9245107c1 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":6},{"id":"e9d11ff9d570c70b1717eacd1a4eed7a8bf792cb","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-10-01 19:12:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e661d674017d9c78b920b696932e9a481a8e8971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 21:04:58.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d943c261e84d459a8f795b20fab424a2\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/80681c7e4f80446f91f88c00c3a24e86 : SUCCESS in 1h 51m 05s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/f801d2c690ba4f4c98390e1fcde109cd : FAILURE in 37m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c8632f1fc0a4e9d82bb232afcf4ec66 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":6},{"id":"84fb14559df43a9c04bbfb56af5bef5154229ef0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-10-04 14:06:12.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"095df9c09464016e9523a9efaa5e5947a583c0d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 16:02:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66550a7825d54464afb2c64417db8ad4\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/037ebcefbb6e4c0189af1db1d7cb7bcc : SUCCESS in 1h 55m 13s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/8429b418312b4b748320b6b9c5cb6fef : FAILURE in 46m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647be29eb8364636bfeed338e628b202 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":7},{"id":"a394bf7f3a78b62121e90d8ca5c60b43b29d2203","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-07 11:54:56.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e06f5f98a6c3f58ac003f9df4337cacb19f91722","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-10-07 16:44:09.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"bd4c2f5359bf88079ac21fffb1b5f5197e2be1a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 19:16:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3891e750bf244a88fd4b7ef3efaab5d\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/d90a05ad88f94a52b84345d3cfc21612 : SUCCESS in 2h 04m 18s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/8f9100a4b0dc45a29e4a926e4adc0aac : SUCCESS in 37m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d78577bc12d84752974eb185bfd94ede : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":8},{"id":"ec296d6f1041b12678c1aaccb05a4e6f1fd243df","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-10-07 19:18:51.000000000","message":"Patch Set 8:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":8},{"id":"cdea94032d57a94af5cee6ba98344ccc7a30a910","author":{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"},"date":"2024-10-07 20:00:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"40ddfc6fc1102c5ebad8646dd955859cd4286e8a","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-10-07 20:02:57.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4633e3f000b974eabfa3a1e9ca7e2d93018b6f22","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-10-07 20:04:59.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"cd3f46082b92820b2130acb8fac459fda0d7a670","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-10-08 00:14:25.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"01d4a473c2af0b517f65e7e3c8039e8ba82e93ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 00:15:04.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6647b738c2543d2efb98dcac1d87605c28902d1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 02:03:38.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab36130c7ac94f8c918cb5a5c7a3579a\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/3f88302d778e4c69b220907690767986 : SUCCESS in 1h 46m 37s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/b7c3c572a8084f58937dc7bce0c0e2b1 : SUCCESS in 47m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16a0796149d04094979b9da8b7f3ce9c : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":8},{"id":"b79ff47185849278016273fa74002c1af215a480","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 02:03:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"261de3a03640287797243bcdccd971e1f342e50b","revisions":{"c7c87edf552249751d4f4d76eeb18affa35e9ca2":{"kind":"REWORK","_number":1,"created":"2024-08-29 15:51:26.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/927525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/927525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/1"}}},"commit":{"parents":[{"commit":"2a09062deb5870f1d8e65afafca83ae3549ad5f7","subject":"Enable tests for tls in whitebox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2a09062deb5870f1d8e65afafca83ae3549ad5f7"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-08-29 15:44:49.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-08-29 15:44:53.000000000","tz":-240},"subject":"Update accessing nova compute logs for RBD check","message":"Update accessing nova compute logs for RBD check\n\nNova compute logs are no longer accessible via accessing the\nnova_compute container downstream. Instead the nova service logs are\navailable directly on the host via journalctl. Updated LogParserClient\nto conditionally access nova service logs either via host or container.\n\nChange-Id: I2cf0ded367e85fca1e49a2cab48adb23b1ed9807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/c7c87edf552249751d4f4d76eeb18affa35e9ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/c7c87edf552249751d4f4d76eeb18affa35e9ca2"}]},"branch":"refs/heads/master"},"0d3ccecd85c2b28aaa0a02aa094cb6e4a61bdba2":{"kind":"REWORK","_number":2,"created":"2024-08-29 16:33:59.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/927525/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/927525/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/2"}}},"commit":{"parents":[{"commit":"2a09062deb5870f1d8e65afafca83ae3549ad5f7","subject":"Enable tests for tls in whitebox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2a09062deb5870f1d8e65afafca83ae3549ad5f7"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-08-29 15:44:49.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-08-29 16:33:50.000000000","tz":-240},"subject":"Update accessing nova compute logs for RBD check","message":"Update accessing nova compute logs for RBD check\n\nNova compute logs are no longer accessible via accessing the\nnova_compute container downstream. Instead the nova service logs are\navailable directly on the host via journalctl. Updated LogParserClient\nto conditionally access nova service logs either via host or container.\n\nChange-Id: I2cf0ded367e85fca1e49a2cab48adb23b1ed9807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/0d3ccecd85c2b28aaa0a02aa094cb6e4a61bdba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/0d3ccecd85c2b28aaa0a02aa094cb6e4a61bdba2"}]},"branch":"refs/heads/master"},"d86cabff1ce6325b1003e94f6b1589b1281097f8":{"kind":"REWORK","_number":3,"created":"2024-09-24 16:50:26.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/927525/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/927525/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/3"}}},"commit":{"parents":[{"commit":"2a09062deb5870f1d8e65afafca83ae3549ad5f7","subject":"Enable tests for tls in whitebox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2a09062deb5870f1d8e65afafca83ae3549ad5f7"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-08-29 15:44:49.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-09-24 16:50:19.000000000","tz":-240},"subject":"Update accessing nova compute logs for RBD check","message":"Update accessing nova compute logs for RBD check\n\nNova compute logs are no longer accessible via accessing the\nnova_compute container downstream. Instead the nova service logs are\navailable directly on the host via journalctl. Updated LogParserClient\nto conditionally access nova service logs either via host or container.\n\nChange-Id: I2cf0ded367e85fca1e49a2cab48adb23b1ed9807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d86cabff1ce6325b1003e94f6b1589b1281097f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d86cabff1ce6325b1003e94f6b1589b1281097f8"}]},"branch":"refs/heads/master"},"c9f938d94a1c4061a51436e7cf4a8f6a179486dd":{"kind":"REWORK","_number":4,"created":"2024-09-25 18:37:02.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/927525/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/927525/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/4"}}},"commit":{"parents":[{"commit":"2a09062deb5870f1d8e65afafca83ae3549ad5f7","subject":"Enable tests for tls in whitebox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2a09062deb5870f1d8e65afafca83ae3549ad5f7"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-08-29 15:44:49.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-09-25 18:36:56.000000000","tz":-240},"subject":"Update accessing nova compute logs for RBD check","message":"Update accessing nova compute logs for RBD check\n\nNova compute logs are no longer accessible via accessing the\nnova_compute container downstream. Instead the nova service logs are\navailable directly on the host via journalctl. Updated LogParserClient\nto conditionally access nova service logs either via host or container.\n\nChange-Id: I2cf0ded367e85fca1e49a2cab48adb23b1ed9807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/c9f938d94a1c4061a51436e7cf4a8f6a179486dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/c9f938d94a1c4061a51436e7cf4a8f6a179486dd"}]},"branch":"refs/heads/master"},"04943d627e14a4725249c22a37df73e654c1359f":{"kind":"REWORK","_number":5,"created":"2024-09-26 15:22:49.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/927525/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/927525/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/5"}}},"commit":{"parents":[{"commit":"2a09062deb5870f1d8e65afafca83ae3549ad5f7","subject":"Enable tests for tls in whitebox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2a09062deb5870f1d8e65afafca83ae3549ad5f7"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-08-29 15:44:49.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-09-26 15:22:46.000000000","tz":-240},"subject":"Update accessing nova compute logs for RBD check","message":"Update accessing nova compute logs for RBD check\n\nNova compute logs are no longer accessible via accessing the\nnova_compute container downstream. Instead the nova service logs are\navailable directly on the host via journalctl. Updated LogParserClient\nto conditionally access nova service logs either via host or container.\n\nChange-Id: I2cf0ded367e85fca1e49a2cab48adb23b1ed9807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/04943d627e14a4725249c22a37df73e654c1359f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/04943d627e14a4725249c22a37df73e654c1359f"}]},"branch":"refs/heads/master"},"51698629b8b75586d792ab339c16c5368f1890ad":{"kind":"REWORK","_number":6,"created":"2024-09-30 16:25:18.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/927525/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/927525/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/6"}}},"commit":{"parents":[{"commit":"2a09062deb5870f1d8e65afafca83ae3549ad5f7","subject":"Enable tests for tls in whitebox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2a09062deb5870f1d8e65afafca83ae3549ad5f7"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-08-29 15:44:49.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-09-30 16:25:15.000000000","tz":-240},"subject":"Update accessing nova compute logs for RBD check","message":"Update accessing nova compute logs for RBD check\n\nNova compute logs are no longer accessible via accessing the\nnova_compute container downstream. Instead the nova service logs are\navailable directly on the host via journalctl. Updated LogParserClient\nto conditionally access nova service logs either via host or container.\n\nChange-Id: I2cf0ded367e85fca1e49a2cab48adb23b1ed9807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/51698629b8b75586d792ab339c16c5368f1890ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/51698629b8b75586d792ab339c16c5368f1890ad"}]},"branch":"refs/heads/master"},"561515618cb4a28b5b70e9708d78bc7349dbf523":{"kind":"REWORK","_number":7,"created":"2024-10-04 14:06:12.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/927525/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/927525/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/7"}}},"commit":{"parents":[{"commit":"2a09062deb5870f1d8e65afafca83ae3549ad5f7","subject":"Enable tests for tls in whitebox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2a09062deb5870f1d8e65afafca83ae3549ad5f7"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-08-29 15:44:49.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-10-04 14:05:59.000000000","tz":-240},"subject":"Update accessing nova compute logs for RBD check","message":"Update accessing nova compute logs for RBD check\n\nNova compute logs are no longer accessible via accessing the\nnova_compute container downstream. Instead the nova service logs are\navailable directly on the host via journalctl. Updated LogParserClient\nto conditionally access nova service logs either via host or container.\n\nChange-Id: I2cf0ded367e85fca1e49a2cab48adb23b1ed9807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/561515618cb4a28b5b70e9708d78bc7349dbf523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/561515618cb4a28b5b70e9708d78bc7349dbf523"}]},"branch":"refs/heads/master"},"261de3a03640287797243bcdccd971e1f342e50b":{"kind":"REWORK","_number":8,"created":"2024-10-07 16:44:09.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/25/927525/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/25/927525/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/25/927525/8"}}},"commit":{"parents":[{"commit":"2a09062deb5870f1d8e65afafca83ae3549ad5f7","subject":"Enable tests for tls in whitebox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2a09062deb5870f1d8e65afafca83ae3549ad5f7"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-08-29 15:44:49.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-10-07 16:44:01.000000000","tz":-240},"subject":"Update accessing nova compute logs for RBD check","message":"Update accessing nova compute logs for RBD check\n\nNova compute logs are no longer accessible via accessing the\nnova_compute container downstream. Instead the nova service logs are\navailable directly on the host via journalctl. Updated LogParserClient\nto conditionally access nova service logs either via host or container.\n\nChange-Id: I2cf0ded367e85fca1e49a2cab48adb23b1ed9807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/261de3a03640287797243bcdccd971e1f342e50b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/261de3a03640287797243bcdccd971e1f342e50b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
