)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~931717","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~I37c64587bf6373c6d4088db9b0d3bec6948ff4ec","project":"openstack/whitebox-tempest-plugin","branch":"master","topic":"nova-imagebackend-poc","attention_set":{"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2024-10-16 18:18:04.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}}},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2024-10-15 15:32:40.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I37c64587bf6373c6d4088db9b0d3bec6948ff4ec","subject":"Add tests for Nova image multi-backend","status":"NEW","created":"2024-10-08 00:59:59.000000000","updated":"2024-10-16 22:03:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"ce48da38df768364eb7cfeea4bb06fde9db98f7a","_number":931717,"virtual_id_number":931717,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-16 22:03:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-10-08 03:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-15 10:06:09.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"CC"}],"messages":[{"id":"80386ded4de67c6a20c98a45f6e4fe40bc8fde6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-08 00:59:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f257f93a8b0f9df45a7bdc7b0d566349f26e22e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-08 01:45:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"73b19cbb39a6cb8c98b9bf3ced6126f9cd6f5483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 03:39:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0c7399b9ebc47f4afdca15ef53c6695\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/560bfcbaf5424074b54972a19eb168ab : SUCCESS in 1h 52m 34s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/7e6bf78d62084e3cbb19042f51e5b33d : SUCCESS in 28m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/701b34d3068b4040b3e586891f28f8b6 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":2},{"id":"7bbac58d2c7001aad1a464f55e5f18f063baff8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-09 07:19:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3721d4818236d921bf4474328611f4306a7343e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 08:53:51.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/930968 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"692e5bdf4814ddff7dd1715223a8a1136551f464","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-09 08:54:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0aa7438326cbe2a53e9fc5a1aebcbc817d5a6759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 11:17:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8611a72b868c440bad2909d2f062ee99\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/827fd57067834e3080e8ded3dc6e3ad5 : SUCCESS in 1h 59m 18s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/1c4e6e7ed2884135b0f829721d6d27f9 : FAILURE in 51m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65992f462e604c25bb0ff78ea7fb8850 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":3},{"id":"09cfda4edcf14717267410d70af281681fe5b0b1","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-09 18:14:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ce35e2177093dcfdd3696190986b63df63f58aa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 19:20:39.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/930968 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"b8dd191bd75f964acce5f2291f37afc926958d54","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-09 20:15:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b00f48e0b07f237ffa5c34add081cf2057a8c46a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 22:16:14.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ad8dcce6be1465e85a174789da45b41\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/1fab52b3851942a4b30fc55729e10c31 : SUCCESS in 1h 59m 34s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/e0486215a12a44e89ea7fcff513238e3 : FAILURE in 50m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad521bd3d4d34a80a9e44df576e37459 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":3},{"id":"d1f7b8e7c43a5a6955ee97d644e03dadbc9d8ba7","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-15 10:06:09.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"12c261e653b54fc0c9a1dfb053a9ed42d3bd3af7","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-15 15:32:40.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"78435b962daaa24514893301c22b7a2ddaac8341","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-16 07:08:40.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"49b9b45109a0bbbe04959469af042156db801aaf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-16 07:47:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f5ea9b6e8a545335b34bc17e375320ac490beb77","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-16 07:49:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1a9dc714b49ae2a4b58ac2ed3e533051dc6d2b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 09:53:43.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0fb14b9ed4244a88cda4de941d5ac85\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/a67eb7606403498d9fc1873a5831cf37 : SUCCESS in 1h 53m 20s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/3fb5b18f48914fcdbf90be544b14d24b : FAILURE in 56m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623a7bd28bf0471ab6ac07ac97e06538 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":6},{"id":"48b463b9f4fdeefdc82da1ebbf648a111f083869","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-16 15:45:28.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"b10884d766341c3edb4f5a66d554221dc2fb1ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 17:34:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9005492f63a43a2bd04e1c32b80af66\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/3dbe0ccff5dc46bfb242c61f3d9a6798 : SUCCESS in 1h 46m 30s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/67443a4042124fbcab3dc7887b8c2e88 : FAILURE in 47m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e11d2c196dd4e31869024e066de87a8 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":7},{"id":"232c8252a37c17cb81e6b975796f57136c6a86c6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-16 18:14:43.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"2e89c11cd078971d43fc77f5a5909a5d2ccf3eae","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-16 18:18:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8773174747735bf30588c7f43f5c44c1717f981e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-16 19:38:48.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ce48da38df768364eb7cfeea4bb06fde9db98f7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 22:03:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10ec24b85a4540e888ef77de3680cccb\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/37362a2c920b4636a79e4781c4bae573 : SUCCESS in 2h 06m 03s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/caab52f1048e4b88929f55b6ce79ceaf : SUCCESS in 59m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d163ee19ee64e66a794074a9481bda7 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"2b3037d68b9ee6c1655441165cd1b70cb2fd563b","revisions":{"b10fd2973c79e3d7fc7a902ed1087df1b1604feb":{"kind":"REWORK","_number":1,"created":"2024-10-08 00:59:59.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/17/931717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/17/931717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/1"}}},"commit":{"parents":[{"commit":"5cbfbdf9d24978e93794ace7e2eead1fbbb42041","subject":"Merge \"Update shared/dedicated cpu calculations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/5cbfbdf9d24978e93794ace7e2eead1fbbb42041"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-03 21:21:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-08 00:59:58.000000000","tz":0},"subject":"Add test for Nova image multi-backend","message":"Add test for Nova image multi-backend\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/930968\n\nChange-Id: I37c64587bf6373c6d4088db9b0d3bec6948ff4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/b10fd2973c79e3d7fc7a902ed1087df1b1604feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/b10fd2973c79e3d7fc7a902ed1087df1b1604feb"}]},"branch":"refs/heads/master"},"9b39e8fa1ffd67210bc158ed2e1726ff1d330be9":{"kind":"REWORK","_number":2,"created":"2024-10-08 01:45:31.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/17/931717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/17/931717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/2"}}},"commit":{"parents":[{"commit":"5cbfbdf9d24978e93794ace7e2eead1fbbb42041","subject":"Merge \"Update shared/dedicated cpu calculations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/5cbfbdf9d24978e93794ace7e2eead1fbbb42041"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-03 21:21:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-08 01:45:18.000000000","tz":0},"subject":"Add test for Nova image multi-backend","message":"Add test for Nova image multi-backend\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/930968\n\nChange-Id: I37c64587bf6373c6d4088db9b0d3bec6948ff4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/9b39e8fa1ffd67210bc158ed2e1726ff1d330be9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/9b39e8fa1ffd67210bc158ed2e1726ff1d330be9"}]},"branch":"refs/heads/master"},"767fbef3b19f79a99392542614db9127232d8475":{"kind":"REWORK","_number":3,"created":"2024-10-09 07:19:10.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/17/931717/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/17/931717/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/3"}}},"commit":{"parents":[{"commit":"5cbfbdf9d24978e93794ace7e2eead1fbbb42041","subject":"Merge \"Update shared/dedicated cpu calculations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/5cbfbdf9d24978e93794ace7e2eead1fbbb42041"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-03 21:21:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-09 07:18:36.000000000","tz":0},"subject":"Add tests for Nova image multi-backend","message":"Add tests for Nova image multi-backend\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/930968\n\nChange-Id: I37c64587bf6373c6d4088db9b0d3bec6948ff4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/767fbef3b19f79a99392542614db9127232d8475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/767fbef3b19f79a99392542614db9127232d8475"}]},"branch":"refs/heads/master"},"ef170b253449c0c88d3179f9bb30867500220f07":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-16 07:08:40.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/17/931717/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/17/931717/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/4"}}},"commit":{"parents":[{"commit":"bf5f2e573fe9846ddd2a772e277b58b06aed6a24","subject":"Merge \"docs: Add DevStack configuration example to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bf5f2e573fe9846ddd2a772e277b58b06aed6a24"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-03 21:21:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-16 07:08:40.000000000","tz":0},"subject":"Add tests for Nova image multi-backend","message":"Add tests for Nova image multi-backend\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/930968\n\nChange-Id: I37c64587bf6373c6d4088db9b0d3bec6948ff4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ef170b253449c0c88d3179f9bb30867500220f07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/ef170b253449c0c88d3179f9bb30867500220f07"}]},"branch":"refs/heads/master","description":"Rebase"},"e070d63d459062e97573b9295d02a34decd68130":{"kind":"REWORK","_number":5,"created":"2024-10-16 07:47:25.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/17/931717/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/17/931717/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/5"}}},"commit":{"parents":[{"commit":"bf5f2e573fe9846ddd2a772e277b58b06aed6a24","subject":"Merge \"docs: Add DevStack configuration example to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bf5f2e573fe9846ddd2a772e277b58b06aed6a24"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-03 21:21:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-16 07:47:12.000000000","tz":0},"subject":"Add tests for Nova image multi-backend","message":"Add tests for Nova image multi-backend\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/930968\n\nChange-Id: I37c64587bf6373c6d4088db9b0d3bec6948ff4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/e070d63d459062e97573b9295d02a34decd68130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/e070d63d459062e97573b9295d02a34decd68130"}]},"branch":"refs/heads/master"},"9e6169f504ed5bae17523b5f407e43d3e27d18d5":{"kind":"REWORK","_number":6,"created":"2024-10-16 07:49:38.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/17/931717/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/17/931717/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/6"}}},"commit":{"parents":[{"commit":"bf5f2e573fe9846ddd2a772e277b58b06aed6a24","subject":"Merge \"docs: Add DevStack configuration example to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bf5f2e573fe9846ddd2a772e277b58b06aed6a24"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-03 21:21:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-16 07:49:14.000000000","tz":0},"subject":"Add tests for Nova image multi-backend","message":"Add tests for Nova image multi-backend\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/930968\n\nChange-Id: I37c64587bf6373c6d4088db9b0d3bec6948ff4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/9e6169f504ed5bae17523b5f407e43d3e27d18d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/9e6169f504ed5bae17523b5f407e43d3e27d18d5"}]},"branch":"refs/heads/master"},"9cb7bacea7a7f0ab1ddfb59b0d2de05d0a4282cb":{"kind":"REWORK","_number":7,"created":"2024-10-16 15:45:28.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/17/931717/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/17/931717/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/7"}}},"commit":{"parents":[{"commit":"bf5f2e573fe9846ddd2a772e277b58b06aed6a24","subject":"Merge \"docs: Add DevStack configuration example to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bf5f2e573fe9846ddd2a772e277b58b06aed6a24"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-03 21:21:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-16 15:43:12.000000000","tz":0},"subject":"Add tests for Nova image multi-backend","message":"Add tests for Nova image multi-backend\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/930968\n\nChange-Id: I37c64587bf6373c6d4088db9b0d3bec6948ff4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/9cb7bacea7a7f0ab1ddfb59b0d2de05d0a4282cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/9cb7bacea7a7f0ab1ddfb59b0d2de05d0a4282cb"}]},"branch":"refs/heads/master"},"12013997554203ca1d54552d63a11fd29ec954de":{"kind":"REWORK","_number":8,"created":"2024-10-16 18:14:43.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/17/931717/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/17/931717/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/8"}}},"commit":{"parents":[{"commit":"bf5f2e573fe9846ddd2a772e277b58b06aed6a24","subject":"Merge \"docs: Add DevStack configuration example to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bf5f2e573fe9846ddd2a772e277b58b06aed6a24"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-03 21:21:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-16 18:14:34.000000000","tz":0},"subject":"Add tests for Nova image multi-backend","message":"Add tests for Nova image multi-backend\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/930968\n\nChange-Id: I37c64587bf6373c6d4088db9b0d3bec6948ff4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/12013997554203ca1d54552d63a11fd29ec954de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/12013997554203ca1d54552d63a11fd29ec954de"}]},"branch":"refs/heads/master"},"2b3037d68b9ee6c1655441165cd1b70cb2fd563b":{"kind":"REWORK","_number":9,"created":"2024-10-16 19:38:48.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/17/931717/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/17/931717/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/17/931717/9"}}},"commit":{"parents":[{"commit":"bf5f2e573fe9846ddd2a772e277b58b06aed6a24","subject":"Merge \"docs: Add DevStack configuration example to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bf5f2e573fe9846ddd2a772e277b58b06aed6a24"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-03 21:21:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-16 19:38:36.000000000","tz":0},"subject":"Add tests for Nova image multi-backend","message":"Add tests for Nova image multi-backend\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/930968\n\nChange-Id: I37c64587bf6373c6d4088db9b0d3bec6948ff4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2b3037d68b9ee6c1655441165cd1b70cb2fd563b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/2b3037d68b9ee6c1655441165cd1b70cb2fd563b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
