)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~933454","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~I76885987842440f9b690c198e6b8149bdd85f899","project":"openstack/whitebox-tempest-plugin","branch":"master","topic":"libvirt_domain_shutdown","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2024-11-20 01:22:11.000000000","reason":"\u003cGERRIT_ACCOUNT_8864\u003e replied on the change","reason_account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}},"31033":{"account":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"last_update":"2024-11-20 03:34:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I76885987842440f9b690c198e6b8149bdd85f899","subject":"Wait for libvirt domain shutdown correctly","status":"MERGED","created":"2024-10-25 18:33:53.000000000","updated":"2024-11-20 03:34:07.000000000","submitted":"2024-11-20 03:34:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"933454-libvirt_domain_shutdown","meta_rev_id":"a2a434ccc9f02234306ce3c887040ab01445f9ff","_number":933454,"virtual_id_number":933454,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 03:34:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":2,"date":"2024-11-20 01:22:11.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":1,"date":"2024-11-20 01:22:11.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}],"REVIEWER":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-25 21:04:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-28 16:42:28.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"state":"CC"}],"messages":[{"id":"7cd966787d14ae6c606e5964a1e94641ba0fab49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-10-25 18:33:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19f264bf8dce9f860ac125d30e1218e7da60b0a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-10-25 18:36:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93c5eaba287aeda5f3aadecd96e75b59775f7c77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-10-25 18:44:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73fa2110455f3fc7779b41bd6d82c9342015144c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 21:04:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8fea03d54614c46bcfdc94f2e6a437a\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/7f5fdb12b6754b429814d8ad3bf2d28d : FAILURE in 2h 18m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/864281cc549b48ee837e251f7d3b0c0a : SUCCESS in 2m 08s","accounts_in_message":[],"_revision_number":3},{"id":"33a167d383869256b7ca5ea06c00a4bd6f3c24da","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-10-28 16:42:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9f0ce6cc2eba390165fa859ce95a46d99b466965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-10-28 17:00:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"26142fcb32bf1b4947aa0d9d955077d9fbf453d2","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-10-28 17:00:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7787260c0067ebb7b122d3f11e542e4958ab4021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 19:01:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad1be270d60c45eaa163d505c09f5c77\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/a1fd583d2ce444aca9f5f0e7fa21ee8c : FAILURE in 1h 58m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de671697150a4190867716d5fe99d514 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":4},{"id":"596b5816f5041ba7f0b37c7691ba55e00f0b2b64","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-10-29 17:20:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2beaaed6601e76fbb1e5c636e3d97e461b85e340","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2024-11-19 19:36:59.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"73f4664cdc03e25f0bbad7494a19340d8c4792d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 21:50:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc07389bd7a641609d43f4d36918be1c\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/51995c092b584358b7f0a611713774cb : SUCCESS in 2h 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/870789b90fe648f8a05375ff8f18f7e9 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":5},{"id":"ba49ad34de7e23c1ed7fab4d72e8f891abe7a116","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-11-20 01:22:11.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a79375d82b010eec03672f38742f1dc6260c4de0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 01:22:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dfeb3332bd4923b6c718d8057b69beccbee5d790","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 03:34:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11882b59429d4748bb3367dbec6c3159\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/a257d4b7d88e40ab8a1131d426b8f124 : SUCCESS in 2h 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/987f9c49ffd34de084122846642b7fcb : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":5},{"id":"a2a434ccc9f02234306ce3c887040ab01445f9ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 03:34:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bee34dbb867dc3c107f1262f68a997ef7ccff55a","revisions":{"505254ec349401ed007b9d9df6740a1497362152":{"kind":"REWORK","_number":1,"created":"2024-10-25 18:33:53.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/54/933454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/54/933454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/1"}}},"commit":{"parents":[{"commit":"bf5f2e573fe9846ddd2a772e277b58b06aed6a24","subject":"Merge \"docs: Add DevStack configuration example to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bf5f2e573fe9846ddd2a772e277b58b06aed6a24"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-10-25 18:28:58.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-10-25 18:28:58.000000000","tz":-240},"subject":"Wait for libvirt domain shutdown correctly","message":"Wait for libvirt domain shutdown correctly\n\nPreviousy, when shutting down a libvirt domain before evacuation, we\nwould wait until it disappeared from the list of libvirt domains, but\nwe just merrily continued on even if it was still present after 60\nseconds. I suspect the domain remained present, and our next steps\nwere to evacuate the instance and start nova-compute back up again, at\nwhich point it tried to delete evacuated instances, and fail with the\nfollowing:\n\nFile \"/opt/stack/nova/nova/compute/manager.py\", line 839, in _destroy_evacuated_instances\n\u003csnip\u003e\nlibvirt.libvirtError: Failed to terminate process 55183 with SIGKILL: Device or resource busy\n\nThis patch makes us wait for the domain to disappear, and raise an\nerror if it doesn\u0027t disappear in time. It also refactors the shutdown\ncode slightly to avoid an extra API quey to get the server.\n\nChange-Id: I76885987842440f9b690c198e6b8149bdd85f899\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/505254ec349401ed007b9d9df6740a1497362152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/505254ec349401ed007b9d9df6740a1497362152"}]},"branch":"refs/heads/master"},"6fea85e0e55b7d72db9b5dd99a9f7d872ad5ef1e":{"kind":"REWORK","_number":2,"created":"2024-10-25 18:36:55.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/54/933454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/54/933454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/2"}}},"commit":{"parents":[{"commit":"bf5f2e573fe9846ddd2a772e277b58b06aed6a24","subject":"Merge \"docs: Add DevStack configuration example to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bf5f2e573fe9846ddd2a772e277b58b06aed6a24"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-10-25 18:28:58.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-10-25 18:36:52.000000000","tz":-240},"subject":"Wait for libvirt domain shutdown correctly","message":"Wait for libvirt domain shutdown correctly\n\nPreviousy, when shutting down a libvirt domain before evacuation, we\nwould wait until it disappeared from the list of libvirt domains, but\nwe just merrily continued on even if it was still present after 60\nseconds. I suspect the domain remained present, and our next steps\nwere to evacuate the instance and start nova-compute back up again, at\nwhich point it tried to delete evacuated instances, and fail with the\nfollowing:\n\nFile \"/opt/stack/nova/nova/compute/manager.py\", line 839, in _destroy_evacuated_instances\n\u003csnip\u003e\nlibvirt.libvirtError: Failed to terminate process 55183 with SIGKILL: Device or resource busy\n\nThis patch makes us wait for the domain to disappear, and raise an\nerror if it doesn\u0027t disappear in time. It also refactors the shutdown\ncode slightly to avoid an extra API quey to get the server.\n\nChange-Id: I76885987842440f9b690c198e6b8149bdd85f899\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6fea85e0e55b7d72db9b5dd99a9f7d872ad5ef1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/6fea85e0e55b7d72db9b5dd99a9f7d872ad5ef1e"}]},"branch":"refs/heads/master"},"0558ad7a4c90a955e4fc00f3facf12e2b396d06f":{"kind":"REWORK","_number":3,"created":"2024-10-25 18:44:22.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/54/933454/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/54/933454/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/3"}}},"commit":{"parents":[{"commit":"bf5f2e573fe9846ddd2a772e277b58b06aed6a24","subject":"Merge \"docs: Add DevStack configuration example to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bf5f2e573fe9846ddd2a772e277b58b06aed6a24"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-10-25 18:28:58.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-10-25 18:44:19.000000000","tz":-240},"subject":"Wait for libvirt domain shutdown correctly","message":"Wait for libvirt domain shutdown correctly\n\nPreviousy, when shutting down a libvirt domain before evacuation, we\nwould wait until it disappeared from the list of libvirt domains, but\nwe just merrily continued on even if it was still present after 60\nseconds. I suspect the domain remained present, and our next steps\nwere to evacuate the instance and start nova-compute back up again, at\nwhich point it tried to delete evacuated instances, and fail with the\nfollowing:\n\nFile \"/opt/stack/nova/nova/compute/manager.py\", line 839, in _destroy_evacuated_instances\n\u003csnip\u003e\nlibvirt.libvirtError: Failed to terminate process 55183 with SIGKILL: Device or resource busy\n\nThis patch makes us wait for the domain to disappear, and raise an\nerror if it doesn\u0027t disappear in time. It also refactors the shutdown\ncode slightly to avoid an extra API quey to get the server.\n\nChange-Id: I76885987842440f9b690c198e6b8149bdd85f899\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/0558ad7a4c90a955e4fc00f3facf12e2b396d06f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/0558ad7a4c90a955e4fc00f3facf12e2b396d06f"}]},"branch":"refs/heads/master"},"9c10f3ec43e8bdd7c41683979dd911a42c78f833":{"kind":"REWORK","_number":4,"created":"2024-10-28 17:00:46.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/54/933454/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/54/933454/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/4"}}},"commit":{"parents":[{"commit":"bf5f2e573fe9846ddd2a772e277b58b06aed6a24","subject":"Merge \"docs: Add DevStack configuration example to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bf5f2e573fe9846ddd2a772e277b58b06aed6a24"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-10-25 18:28:58.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-10-28 17:00:41.000000000","tz":-240},"subject":"Wait for libvirt domain shutdown correctly","message":"Wait for libvirt domain shutdown correctly\n\nPreviousy, when shutting down a libvirt domain before evacuation, we\nwould wait until it disappeared from the list of libvirt domains, but\nwe just merrily continued on even if it was still present after 60\nseconds. I suspect the domain remained present, and our next steps\nwere to evacuate the instance and start nova-compute back up again, at\nwhich point it tried to delete evacuated instances, and fail with the\nfollowing:\n\nFile \"/opt/stack/nova/nova/compute/manager.py\", line 839, in _destroy_evacuated_instances\n\u003csnip\u003e\nlibvirt.libvirtError: Failed to terminate process 55183 with SIGKILL: Device or resource busy\n\nThis patch makes us wait for the domain to disappear, and raise an\nerror if it doesn\u0027t disappear in time. It also refactors the shutdown\ncode slightly to avoid an extra API quey to get the server.\n\nChange-Id: I76885987842440f9b690c198e6b8149bdd85f899\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/9c10f3ec43e8bdd7c41683979dd911a42c78f833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/9c10f3ec43e8bdd7c41683979dd911a42c78f833"}]},"branch":"refs/heads/master"},"bee34dbb867dc3c107f1262f68a997ef7ccff55a":{"kind":"REWORK","_number":5,"created":"2024-11-19 19:36:59.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/54/933454/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/54/933454/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/54/933454/5"}}},"commit":{"parents":[{"commit":"bf5f2e573fe9846ddd2a772e277b58b06aed6a24","subject":"Merge \"docs: Add DevStack configuration example to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bf5f2e573fe9846ddd2a772e277b58b06aed6a24"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-10-25 18:28:58.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2024-11-19 19:36:54.000000000","tz":-300},"subject":"Wait for libvirt domain shutdown correctly","message":"Wait for libvirt domain shutdown correctly\n\nPreviousy, when shutting down a libvirt domain before evacuation, we\nwould wait until it disappeared from the list of libvirt domains, but\nwe just merrily continued on even if it was still present after 60\nseconds. I suspect the domain remained present, and our next steps\nwere to evacuate the instance and start nova-compute back up again, at\nwhich point it tried to delete evacuated instances, and fail with the\nfollowing:\n\nFile \"/opt/stack/nova/nova/compute/manager.py\", line 839, in _destroy_evacuated_instances\n\u003csnip\u003e\nlibvirt.libvirtError: Failed to terminate process 55183 with SIGKILL: Device or resource busy\n\nThis patch makes us wait for the domain to disappear, and raise an\nerror if it doesn\u0027t disappear in time. It also refactors the shutdown\ncode slightly to avoid an extra API quey to get the server.\n\nChange-Id: I76885987842440f9b690c198e6b8149bdd85f899\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bee34dbb867dc3c107f1262f68a997ef7ccff55a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/bee34dbb867dc3c107f1262f68a997ef7ccff55a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
