)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~942511","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~Ic6ed360ee18e5b000c2305c65e1fb5595a95f196","project":"openstack/whitebox-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2025-04-25 20:49:34.000000000","reason":"Change was submitted"},"31033":{"account":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"last_update":"2025-04-25 17:00:35.000000000","reason":"\u003cGERRIT_ACCOUNT_31033\u003e replied on the change","reason_account":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}}},"hashtags":[],"change_id":"Ic6ed360ee18e5b000c2305c65e1fb5595a95f196","subject":"Disable CPU power management","status":"MERGED","created":"2025-02-22 14:34:58.000000000","updated":"2025-04-25 20:49:34.000000000","submitted":"2025-04-25 20:49:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942511","meta_rev_id":"488894bd7fab9b694f4aafe13ce08be777470563","_number":942511,"virtual_id_number":942511,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-25 20:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"all":[{"value":2,"date":"2025-04-25 18:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"all":[{"value":1,"date":"2025-04-25 18:49:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-22 15:23:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-25 16:45:10.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"state":"REVIEWER"}],"messages":[{"id":"fb819b735da1e9bfbcc2e1ecc38f169a6775398d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2025-02-22 14:34:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35f495f3f65792abcb807bacd4b3f9baaf1f8ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 15:23:43.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09ee01c1c82b44a4ab8df24111833cd6\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/9b1bf3fed0fe414c9cc2b753cd27f481 : FAILURE in 47m 20s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/7123e41c080346aca1a765dc4a67f610 : SUCCESS in 31m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5740834dff84f6ba013ad8ced794978 : FAILURE in 2m 09s","accounts_in_message":[],"_revision_number":1},{"id":"37fb77c3903243e5c2a96a6c04dbb03422017681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2025-02-22 15:41:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7be1be93a34c1fc749acef77876eaf5df491d074","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 18:05:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a711ead49d1f4c07a4dd29ed5c2bdcc4\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/9d0f8981ba3a414daf5d5809dcba2eb2 : SUCCESS in 2h 21m 47s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/81ad7c125d1446a8919cbb9bd9d44218 : SUCCESS in 1h 00m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/679badeed48e4faeaa7bfcbac0ec4f16 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f4d3c246d2692c97637c94315727c88245abd4c9","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 16:45:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fc3c536bd7088709e62c6d934c2e659ae5e9df20","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 16:45:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc446fc3a81b2aabe01d4829ad431e477b23e558","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 16:46:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d36495dd4c014df2c4d788bf1e9209b6f440b5b","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2025-04-25 16:46:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b5d77433d859bdcb2dc25bcffac86a5308ec4c5","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 17:00:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe91c6a8ad01382bd44f916380e7066dbb72bd45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 18:48:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d09ef4d447940d9a67aaf7dabbaee8f\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/444e3f572f3a45ae8b935086ef0691f5 : SUCCESS in 2h 01m 12s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/47bb5a67a34b43e69c88a300d594065f : SUCCESS in 49m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2f6005f970548bd81eed99ec31cf8b9 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":2},{"id":"1fbb7d7d8a59fef66f106fe69e4c1792503fa01c","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 18:49:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f966b8b0e25b37372fd7a4ed16d190c939a4144e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 18:49:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f537646b894e2d59bb7bd109db8859d4763c2cea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 20:49:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f62d5faa62a4fdbaca8074580ac67df\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/ac184073a4d14e5c9f59b867b202abe2 : SUCCESS in 1h 59m 28s\n- whitebox-devstack-ceph-multinode https://zuul.opendev.org/t/openstack/build/dd4048b459214ddf854cdc1dd5d16c91 : SUCCESS in 43m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bab7341575b040048ad125d829f91a30 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":2},{"id":"488894bd7fab9b694f4aafe13ce08be777470563","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 20:49:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"36d5633f67656b57f7f089e48f93ce60025a1ed3","revisions":{"d05fb37e350a35ade7e45997095d01862b700907":{"kind":"REWORK","_number":1,"created":"2025-02-22 14:34:58.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/11/942511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/11/942511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/11/942511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/11/942511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/11/942511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/11/942511/1"}}},"commit":{"parents":[{"commit":"a464d48e5e517ac8e4d04ca72b25042067abcb31","subject":"Update string format to address PEP8 E231","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a464d48e5e517ac8e4d04ca72b25042067abcb31"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2025-02-22 14:14:07.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2025-02-22 14:15:11.000000000","tz":-300},"subject":"Disable CPU power management","message":"Disable CPU power management\n\nUntil we fix the bugs, disable it to get clean runs, and fix our tests\nto correctly skip if the feature is not available.\n\nChange-Id: Ic6ed360ee18e5b000c2305c65e1fb5595a95f196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d05fb37e350a35ade7e45997095d01862b700907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/d05fb37e350a35ade7e45997095d01862b700907"}]},"branch":"refs/heads/master"},"36d5633f67656b57f7f089e48f93ce60025a1ed3":{"kind":"REWORK","_number":2,"created":"2025-02-22 15:41:21.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/11/942511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/11/942511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/11/942511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/11/942511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/11/942511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/11/942511/2"}}},"commit":{"parents":[{"commit":"a464d48e5e517ac8e4d04ca72b25042067abcb31","subject":"Update string format to address PEP8 E231","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a464d48e5e517ac8e4d04ca72b25042067abcb31"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2025-02-22 14:14:07.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2025-02-22 15:41:11.000000000","tz":-300},"subject":"Disable CPU power management","message":"Disable CPU power management\n\nUntil we fix the bugs, disable it to get clean runs, and fix our tests\nto correctly skip if the feature is not available.\n\nChange-Id: Ic6ed360ee18e5b000c2305c65e1fb5595a95f196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/36d5633f67656b57f7f089e48f93ce60025a1ed3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/36d5633f67656b57f7f089e48f93ce60025a1ed3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
