)]}'
{"id":"openstack%2Fwhitebox-tempest-plugin~947855","triplet_id":"openstack%2Fwhitebox-tempest-plugin~master~I412f55327d475ee5bc7a9a34c9e5012d1de3c3e0","project":"openstack/whitebox-tempest-plugin","branch":"master","topic":"vtpm_invalid_ul_update","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2025-05-13 16:59:14.000000000","reason":"\u003cGERRIT_ACCOUNT_8864\u003e replied on the change","reason_account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}},"35153":{"account":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"last_update":"2025-05-13 21:37:19.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-05-13 21:37:19.000000000","reason":"Change was submitted"},"34309":{"account":{"_account_id":34309,"name":"Sreelakshmi Menon Kovili","display_name":"Sree","email":"skovili@redhat.com","username":"sreekovili"},"last_update":"2025-05-13 21:37:19.000000000","reason":"Change was submitted"},"31033":{"account":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"last_update":"2025-05-13 21:37:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I412f55327d475ee5bc7a9a34c9e5012d1de3c3e0","subject":"Update build exception when using unified limits","status":"MERGED","created":"2025-04-22 17:00:12.000000000","updated":"2025-05-13 21:37:19.000000000","submitted":"2025-05-13 21:37:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"947855-vtpm_invalid_ul_update","meta_rev_id":"762e5dee677ff7950478b943fb4e0729f98a3a3d","_number":947855,"virtual_id_number":947855,"owner":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-13 21:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":2,"date":"2025-05-13 16:59:14.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":1,"date":"2025-05-13 16:59:14.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"date":"2025-05-08 15:44:38.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},{"_account_id":34309,"name":"Sreelakshmi Menon Kovili","display_name":"Sree","email":"skovili@redhat.com","username":"sreekovili"}],"reviewers":{"REVIEWER":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"_account_id":34309,"name":"Sreelakshmi Menon Kovili","display_name":"Sree","email":"skovili@redhat.com","username":"sreekovili"},{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-22 17:34:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-25 13:34:10.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"state":"REVIEWER"},{"updated":"2025-04-25 13:34:10.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":34309,"name":"Sreelakshmi Menon Kovili","display_name":"Sree","email":"skovili@redhat.com","username":"sreekovili"},"state":"REVIEWER"},{"updated":"2025-04-25 13:34:10.000000000","updated_by":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-04-25 16:41:53.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"CC"},{"updated":"2025-04-25 16:43:31.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"}],"messages":[{"id":"3b463452dd70648cfb48768e4205a24dd1170ef2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-22 17:00:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90717cabddbd42f461f83b31bf651611ffcde9aa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-22 17:01:44.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e55d2c441b094829a2ceeb5f7eff528d71fc2d47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 17:34:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b640b0b0fd64f4fa202dedd667ca8ad\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/fb0edac7aec44317b2381621c8d833b9 : FAILURE in 31m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95b5fce3e197444b9bc42e5b8eddebec : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":2},{"id":"7663473f45aac73155928a62caac9e8d596a60ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-22 18:22:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"789942a682c4e9e7b32e9b7b360cf4fa266000c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 20:46:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d84bf408080e45c1a44b4e3cfd22e96a\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/835b8e424d424ae8883ac9a834c2a3d7 : SUCCESS in 2h 22m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad1764f485245a08fda9e1b2482d288 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"91e3014209f5a5d41aa09a33ed75722b7cca991a","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 13:34:10.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"6d9f382b19b6f24fa85dac337dbbf19484a57dd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 13:52:55.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"acc3687e19a8eb650c0fc9921bf1bb8ae19d4467","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-25 14:02:25.000000000","message":"Patch Set 3: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fd569c8ce0fb2cc88e03c44dd7181714bf629396","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-25 14:04:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f88c63b78119c5acc7a345956bbe7d9f0c24b46b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 14:39:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35cefec57c7b451181635003f0641f45\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/4923b0dfd9fe48f998d7f93f68dafa8c : FAILURE in 37m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e08d12f1b12a453ca1a6055e3a6f43b3 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":4},{"id":"bce275f21fc101da7cd6c9c632028870891821ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 14:55:30.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ef5caa274ce56b4f6854471428eb3d4b32bd4dec","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 14:56:00.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"da26dbb6d671ed1109414c695a703bf5a732bd8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 14:57:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2386b1c9fdce886e3862750412b8e8327c1221db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 15:01:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f564e2be52e23df6a71c07ebf8624b440ad55008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 15:38:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dd370cb4745466db5bbeccea69aad62\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/494a98a0b93449c4b77273ed00d19250 : FAILURE in 35m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be94545e21474ad791cfa25471932bc3 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":7},{"id":"5f41d4988f4a209c4eceafe08929b865a4087d0b","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-25 16:25:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"56b070129991a2156d80d2d03942983d1c9df16e","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2025-04-25 16:41:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"30f486be778052061cf9ade6615703dc804916a7","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2025-04-25 16:43:31.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"58cde891978932ed28891c99352b6fc621a1aabf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 18:28:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf08f3b1a1ca4a8da5b9a1a0225d9204\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/92108fd35dce47e9ae43f41d418df289 : SUCCESS in 2h 02m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad6ea514839740a2ac0ef57a09f2e4ed : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":7},{"id":"eedd4e436ef92e9ccff81d9cbd8aa04bb5916f32","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2025-04-28 15:41:33.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"6f70d0b6963dfce87c0af46792d27d3eae41ac2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 15:41:57.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2087dea9d2258b284899620b395fb8ba01d86b7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 17:50:55.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c584d3a2cc274f3a8a237a23529690b8\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/cd8fd83589cc41159f3a57b640e321fc : FAILURE in 2h 02m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6fd9762be484ad8a28fac174dd205b1 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":7},{"id":"f19634a21ac57e59c76ae06ce0ae78014343ca47","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-28 18:11:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0d4360448ea49473593939e6526c826608e3d1ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:24:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b14ff02321b7448badeef2e67c410ca6\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/4b28b67e8cc94e59baed60431773712d : FAILURE in 2h 04m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc80bc07f4b4d0ba64701b5bb50612e : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":7},{"id":"8667fe76305102d9d4be03a69ed5316311a8a2c4","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-29 15:34:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4242c56a1c6f829638b002ca1fddcdef2b461050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 17:41:49.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bf5aa1b27344c788d5d9bf43ee7ca4b\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/5b0b3af77ca84687a34e19a90dc2f296 : FAILURE in 2h 06m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb3d9a6e51234c5487fe94c45c8b61cc : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":7},{"id":"2bd7cb49ae3261d961f6b9a70ac2eae81b8935c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"real_author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2025-04-29 18:24:12.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_31033\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}],"_revision_number":8},{"id":"8ee94de8f7dcbb25794c1cb771334680cef69447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 20:26:39.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/810af27d5f7e46f58adc54ea51cbb386\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/7cf6775895d847848731bfbb559cf8d7 : FAILURE in 2h 01m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf93288b17de4f95be7d640920e29f0d : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":8},{"id":"0fab6504345a1a93f568b75f6c659fdc2746ee13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-04-30 14:48:45.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent tree, and commit message as Patch Set 8.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"0d1299ad3942a922905dfd9ba7e4c5a7a9978831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 16:51:19.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82cf23181ef94eadbd12dfac40f2cd6d\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/07656d8d88c24f05adb27ab50285ddc7 : FAILURE in 2h 01m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad6b85d0830b44c5ad900e13eb86ff38 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":9},{"id":"86652d36c2e47e995c40db9fbe31186b30cc898a","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-05-05 15:13:23.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d85e253f6ee2223debc3a6cef8dd82ca5f9c9712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 17:13:25.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4fc4333cdce4155978ce519d34ffce1\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/ae6b5d03fba645dda941eff2ed0339de : FAILURE in 1h 58m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7890016d0b414b1687ec36588fc16672 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":9},{"id":"0af30a776b562d9d1d490279816529864f60a15e","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-05-07 23:20:24.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6cc4e1c50605e55110fa6fba51658f562ee47938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 00:01:25.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/114129d250cd4c82a13864f44b662764\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/7216e765717d44979ac26881080f21b8 : FAILURE in 40m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/105ce2caba124019bd5113fc9806816b : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":9},{"id":"c5dcd7ea7495236ff54be1c1d04cf54acf0a3726","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-08 00:24:31.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9e5f648bb59c02c046404f1108801aca877744ec","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-08 00:25:21.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8955fd6ccdd2ed847f75a7c8e2c8d7e725e7cfe8","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-05-08 13:39:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c387639e1d5a67cfbb9d2636896eb63d6dc79fe0","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-05-08 13:39:43.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"11c54894d7743b376adff7b233df8221c5d8ff20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 15:43:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d3407e30dfc464398953cd215b47e46\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/e0b7f0951e394a6d80e9183ba557fa8c : SUCCESS in 2h 01m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42df6ba31e80489daa962e76465c8c2e : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":9},{"id":"196190862e37093ce1d7a9e6b600bc7c075f9827","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-05-08 15:43:48.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"5966e31e2fa93808f568768226c430ff2dafad45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 15:44:09.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"1f671699cea102df622f606b9d00ce4ec90e83af","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2025-05-08 15:44:38.000000000","message":"Patch Set 9: -Workflow","accounts_in_message":[],"_revision_number":9},{"id":"cb0d14cb94d2cd3e0155c91e46051026190dd296","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 15:44:39.000000000","message":"Patch Set 9: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":9},{"id":"2bce2ed6c5c93e90dcb5860c4ff395832f10f923","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2025-05-13 16:59:14.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"c678d47e58c387b8f827c99b68eb17a776cbef3c","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2025-05-13 16:59:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"13b06cea157361b2ebac6f7cb12f2a09f34983e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 19:11:55.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a739d8b84b884173a44456a13eab5589\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/25b342cad43f40b58fef36f285ae3e6e : SUCCESS in 2h 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd5676dffc464bfca03f986e3247b020 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":9},{"id":"b151db2faa96752a32a8247b357d03e3f1937981","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 19:12:23.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4b7e89276bb8f53e1811ef9c9035cd78ae30473e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 21:37:19.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c59930c953e04d6a99302e714aeda5d8\n\n- whitebox-devstack-multinode https://zuul.opendev.org/t/openstack/build/c3e2e7c1956649b884aadc839c0d1b25 : SUCCESS in 2h 23m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60e30dba8d8449f1b3640415a3aecbef : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":9},{"id":"762e5dee677ff7950478b943fb4e0729f98a3a3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 21:37:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"e056798f095f7758ed7221fcb8c9a25b6a33d692","revisions":{"a3197906ec02101ee0d405b5b3b38e9caa282f82":{"kind":"REWORK","_number":1,"created":"2025-04-22 17:00:12.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/55/947855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/55/947855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/1"}}},"commit":{"parents":[{"commit":"a464d48e5e517ac8e4d04ca72b25042067abcb31","subject":"Update string format to address PEP8 E231","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a464d48e5e517ac8e4d04ca72b25042067abcb31"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 17:00:02.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 17:00:02.000000000","tz":-240},"subject":"Update build exception when using unified limits","message":"Update build exception when using unified limits\n\nChange-Id: I412f55327d475ee5bc7a9a34c9e5012d1de3c3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a3197906ec02101ee0d405b5b3b38e9caa282f82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a3197906ec02101ee0d405b5b3b38e9caa282f82"}]},"branch":"refs/heads/master"},"a9e5a0de58735d3c684ccc2d1f7e5239a36cc788":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-04-22 17:01:44.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/55/947855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/55/947855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/2"}}},"commit":{"parents":[{"commit":"a464d48e5e517ac8e4d04ca72b25042067abcb31","subject":"Update string format to address PEP8 E231","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a464d48e5e517ac8e4d04ca72b25042067abcb31"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 17:00:02.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 17:00:18.000000000","tz":-240},"subject":"Update build exception when using unified limits","message":"Update build exception when using unified limits\n\nWhen unified limits is enabled invalid vtpm models and versions will be\ncaught by Nova-Api and throw a server fault. Updating the negative vTPM\ntest to check for different exceptions based on if unified limits has\nbeen enabled.\n\nChange-Id: I412f55327d475ee5bc7a9a34c9e5012d1de3c3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a9e5a0de58735d3c684ccc2d1f7e5239a36cc788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a9e5a0de58735d3c684ccc2d1f7e5239a36cc788"}]},"branch":"refs/heads/master"},"91239bc2926206647c081479b5231bdee8e0b35c":{"kind":"REWORK","_number":3,"created":"2025-04-22 18:22:51.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/55/947855/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/55/947855/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/3"}}},"commit":{"parents":[{"commit":"a464d48e5e517ac8e4d04ca72b25042067abcb31","subject":"Update string format to address PEP8 E231","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a464d48e5e517ac8e4d04ca72b25042067abcb31"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 17:00:02.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 18:22:47.000000000","tz":-240},"subject":"Update build exception when using unified limits","message":"Update build exception when using unified limits\n\nWhen unified limits is enabled invalid vtpm models and versions will be\ncaught by Nova-Api and throw a server fault. Updating the negative vTPM\ntest to check for different exceptions based on if unified limits has\nbeen enabled.\n\nChange-Id: I412f55327d475ee5bc7a9a34c9e5012d1de3c3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/91239bc2926206647c081479b5231bdee8e0b35c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/91239bc2926206647c081479b5231bdee8e0b35c"}]},"branch":"refs/heads/master"},"3f5839e759df33cd9813ed8a0ac691a98ed4c122":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-04-25 13:52:55.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/55/947855/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/55/947855/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/4"}}},"commit":{"parents":[{"commit":"a464d48e5e517ac8e4d04ca72b25042067abcb31","subject":"Update string format to address PEP8 E231","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a464d48e5e517ac8e4d04ca72b25042067abcb31"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 17:00:02.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-25 13:46:02.000000000","tz":-240},"subject":"Update build exception when using unified limits","message":"Update build exception when using unified limits\n\nWhen unified limits is enabled the unsupported vtpm models/version being\ntested downstream in test_invalid_model_version_creation is now throwing\na new exception. This is because Unified Limits will call\nfrom_request_spec and confirm the flavor extra_specs are valid earlier\nin the path. The resulting error now throws a ServerFault from Nova-Api.\nUpdating the test so that when Unified Limits are configured, it will\ncheck for ServerFault when trying to build with unsupported flavor\nspecs, but still check for BuildExcpetion if Unified Limits have not\nbeen enabled.\n\nChange-Id: I412f55327d475ee5bc7a9a34c9e5012d1de3c3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/3f5839e759df33cd9813ed8a0ac691a98ed4c122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/3f5839e759df33cd9813ed8a0ac691a98ed4c122"}]},"branch":"refs/heads/master"},"31d3092232dbdab271a88563194cc034c7325e79":{"kind":"REWORK","_number":5,"created":"2025-04-25 14:55:30.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/55/947855/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/55/947855/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/5"}}},"commit":{"parents":[{"commit":"a464d48e5e517ac8e4d04ca72b25042067abcb31","subject":"Update string format to address PEP8 E231","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a464d48e5e517ac8e4d04ca72b25042067abcb31"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 17:00:02.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-25 14:49:45.000000000","tz":-240},"subject":"Update build exception when using unified limits","message":"Update build exception when using unified limits\n\nWhen Unified Limits are enabled, the invalid vtpm models being tested\ndownstream in test_invalid_model_version_creation is now throwing a new\nexception. This is because Unified Limits will call from_request_spec\nand confirm the flavor extra_specs are valid earlier in the path. The\nresulting error now throws a ServerFault from Nova-Api.  Updating the\ntest so that when Unified Limits are configured, it will check for\nServerFault when trying to build with invalid flavor specs, but still\ncheck for BuildExcpetion if Unified Limits have not been enabled.\n\nChange-Id: I412f55327d475ee5bc7a9a34c9e5012d1de3c3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/31d3092232dbdab271a88563194cc034c7325e79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/31d3092232dbdab271a88563194cc034c7325e79"}]},"branch":"refs/heads/master"},"5ef91c66f2e27592fde7d45d5750f333249a1b2e":{"kind":"REWORK","_number":6,"created":"2025-04-25 14:57:58.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/55/947855/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/55/947855/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/6"}}},"commit":{"parents":[{"commit":"a464d48e5e517ac8e4d04ca72b25042067abcb31","subject":"Update string format to address PEP8 E231","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a464d48e5e517ac8e4d04ca72b25042067abcb31"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 17:00:02.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-25 14:57:54.000000000","tz":-240},"subject":"Update build exception when using unified limits","message":"Update build exception when using unified limits\n\nWhen Unified Limits are enabled, the invalid vtpm models being tested\ndownstream in test_invalid_model_version_creation is now throwing a new\nexception. This is because Unified Limits will call from_request_spec\nand confirm the flavor extra_specs are valid earlier in the path. The\nresulting error now throws a ServerFault from Nova-Api.  Updating the\ntest so that when Unified Limits are configured, it will check for\nServerFault when trying to build with invalid flavor specs, but still\ncheck for BuildExcpetion if Unified Limits have not been enabled.\n\nChange-Id: I412f55327d475ee5bc7a9a34c9e5012d1de3c3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/5ef91c66f2e27592fde7d45d5750f333249a1b2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/5ef91c66f2e27592fde7d45d5750f333249a1b2e"}]},"branch":"refs/heads/master"},"8cc526424303eb3ae540bf0964e69d66206f0972":{"kind":"REWORK","_number":7,"created":"2025-04-25 15:01:29.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/55/947855/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/55/947855/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/7"}}},"commit":{"parents":[{"commit":"a464d48e5e517ac8e4d04ca72b25042067abcb31","subject":"Update string format to address PEP8 E231","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/a464d48e5e517ac8e4d04ca72b25042067abcb31"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 17:00:02.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-25 15:01:26.000000000","tz":-240},"subject":"Update build exception when using unified limits","message":"Update build exception when using unified limits\n\nWhen Unified Limits are enabled, the invalid vtpm models being tested\ndownstream in test_invalid_model_version_creation is now throwing a new\nexception. This is because Unified Limits will call from_request_spec\nand confirm the flavor extra_specs are valid earlier in the path. The\nresulting error now throws a ServerFault from Nova-Api.  Updating the\ntest so that when Unified Limits are configured, it will check for\nServerFault when trying to build with invalid flavor specs, but still\ncheck for BuildExcpetion if Unified Limits have not been enabled.\n\nChange-Id: I412f55327d475ee5bc7a9a34c9e5012d1de3c3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/8cc526424303eb3ae540bf0964e69d66206f0972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/8cc526424303eb3ae540bf0964e69d66206f0972"}]},"branch":"refs/heads/master"},"23cf858fb551fbaaa09ce29beb5f7953f09fec83":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-04-29 18:24:12.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"real_uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/55/947855/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/55/947855/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/8"}}},"commit":{"parents":[{"commit":"36d5633f67656b57f7f089e48f93ce60025a1ed3","subject":"Disable CPU power management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/36d5633f67656b57f7f089e48f93ce60025a1ed3"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 17:00:02.000000000","tz":-240},"committer":{"name":"James Parker","email":"jparker@redhat.com","date":"2025-04-29 18:24:12.000000000","tz":0},"subject":"Update build exception when using unified limits","message":"Update build exception when using unified limits\n\nWhen Unified Limits are enabled, the invalid vtpm models being tested\ndownstream in test_invalid_model_version_creation is now throwing a new\nexception. This is because Unified Limits will call from_request_spec\nand confirm the flavor extra_specs are valid earlier in the path. The\nresulting error now throws a ServerFault from Nova-Api.  Updating the\ntest so that when Unified Limits are configured, it will check for\nServerFault when trying to build with invalid flavor specs, but still\ncheck for BuildExcpetion if Unified Limits have not been enabled.\n\nChange-Id: I412f55327d475ee5bc7a9a34c9e5012d1de3c3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/23cf858fb551fbaaa09ce29beb5f7953f09fec83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/23cf858fb551fbaaa09ce29beb5f7953f09fec83"}]},"branch":"refs/heads/master","description":"Rebase"},"e056798f095f7758ed7221fcb8c9a25b6a33d692":{"kind":"NO_CHANGE","_number":9,"created":"2025-04-30 14:48:45.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/55/947855/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/whitebox-tempest-plugin","ref":"refs/changes/55/947855/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/whitebox-tempest-plugin refs/changes/55/947855/9"}}},"commit":{"parents":[{"commit":"36d5633f67656b57f7f089e48f93ce60025a1ed3","subject":"Disable CPU power management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/36d5633f67656b57f7f089e48f93ce60025a1ed3"}]}],"author":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-22 17:00:02.000000000","tz":-240},"committer":{"name":"jamepark4","email":"jparker@redhat.com","date":"2025-04-30 14:48:38.000000000","tz":-240},"subject":"Update build exception when using unified limits","message":"Update build exception when using unified limits\n\nWhen Unified Limits are enabled, the invalid vtpm models being tested\ndownstream in test_invalid_model_version_creation is now throwing a new\nexception. This is because Unified Limits will call from_request_spec\nand confirm the flavor extra_specs are valid earlier in the path. The\nresulting error now throws a ServerFault from Nova-Api.  Updating the\ntest so that when Unified Limits are configured, it will check for\nServerFault when trying to build with invalid flavor specs, but still\ncheck for BuildExcpetion if Unified Limits have not been enabled.\n\nChange-Id: I412f55327d475ee5bc7a9a34c9e5012d1de3c3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/e056798f095f7758ed7221fcb8c9a25b6a33d692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/whitebox-tempest-plugin/commit/e056798f095f7758ed7221fcb8c9a25b6a33d692"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
