)]}'
{"id":"openstack%2Fxstatic-bootstrap-scss~788168","triplet_id":"openstack%2Fxstatic-bootstrap-scss~master~I1c884af39871d1393781ec2153c925a876feb53e","project":"openstack/xstatic-bootstrap-scss","branch":"master","hashtags":[],"change_id":"I1c884af39871d1393781ec2153c925a876feb53e","subject":"Use py3 as the default runtime for tox","status":"NEW","created":"2021-04-27 06:41:58.000000000","updated":"2021-04-28 11:55:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8afbce7f44081019183e006dba270446776dcc88","_number":788168,"virtual_id_number":788168,"owner":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-27 06:42:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-1,"date":"2021-04-28 11:55:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 06:42:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 11:55:09.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"15130e8b1eb017515cf61204e735bfff68c8b94d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2021-04-27 06:41:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06c6654ecf09b07781f627da9c973de292adc212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 06:42:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/ac2cb0b5e73743e9bd56af01e0accc4b : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"8afbce7f44081019183e006dba270446776dcc88","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-04-28 11:55:09.000000000","message":"Patch Set 1: Code-Review-1\n\nCould you please add more info. about why we really need this change.\nAfter your change if you run \u0027tox -e py3\u0027 then you use default python3 on the\nuser platform, not the one requested for the testing.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b9f2cf4a63949d42cf6181a54bc1a9836f54652","revisions":{"6b9f2cf4a63949d42cf6181a54bc1a9836f54652":{"kind":"REWORK","_number":1,"created":"2021-04-27 06:41:58.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/68/788168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/xstatic-bootstrap-scss","ref":"refs/changes/68/788168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/xstatic-bootstrap-scss refs/changes/68/788168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/xstatic-bootstrap-scss refs/changes/68/788168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/xstatic-bootstrap-scss refs/changes/68/788168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/xstatic-bootstrap-scss refs/changes/68/788168/1"}}},"commit":{"parents":[{"commit":"1ba0b953862e953a8af9fc8a0f1c9ff372b169cf","subject":"Upgrade to 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-bootstrap-scss/commit/1ba0b953862e953a8af9fc8a0f1c9ff372b169cf"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2021-04-27 06:41:35.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2021-04-27 06:41:53.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nChange-Id: I1c884af39871d1393781ec2153c925a876feb53e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-bootstrap-scss/commit/6b9f2cf4a63949d42cf6181a54bc1a9836f54652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-bootstrap-scss/commit/6b9f2cf4a63949d42cf6181a54bc1a9836f54652"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
