)]}'
{"id":"openstack%2Fxstatic-jquery.tablesorter~717126","triplet_id":"openstack%2Fxstatic-jquery.tablesorter~master~Ibe7de2ab94752d0fc3a829c93da3e7f9bede8aa6","project":"openstack/xstatic-jquery.tablesorter","branch":"master","topic":"release-preparation","hashtags":[],"change_id":"Ibe7de2ab94752d0fc3a829c93da3e7f9bede8aa6","subject":"Prepare for relase using OpenStack release process","status":"MERGED","created":"2020-04-02 21:17:30.000000000","updated":"2020-04-07 13:35:14.000000000","submitted":"2020-04-07 13:35:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717126-1586266514784-98f64a33","meta_rev_id":"cd46aa66bfde26813bc84661e8edb0045922f50f","_number":717126,"virtual_id_number":717126,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 13:35:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2020-04-07 13:35:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-04-07 10:00:40.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2020-04-07 13:35:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 10:00:40.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-04-07 13:35:00.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-04-07 13:35:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e09c19bdecf83703f8ef670d3347876f1f5d487","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-04-02 21:17:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c68e9ee1e13510d36dd89fce5e51ac26c2ba503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 21:17:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/82ff49e8cd484929ae612f3ed246579a : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"a2f4e7a82b9f67f019c71cc1350d48dba9817a21","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-07 10:00:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cb47b9c70e2c470b593602d1f628c5c0176838ce","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-04-07 13:35:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"41ce7c6b54962ea6de6aff9569f9c625ef06640b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 13:35:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e0fc5da448b3343013539905171eea1f9bfe9579","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 13:35:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/5cfb2e0d99b04fcfb60be5a96ddd8318 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"8dd50a288b6672b03a3374eb23bfc73858e0f313","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 13:35:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09a700b565e3fa8696d8bfd1fc9059e1979f9f79","revisions":{"09a700b565e3fa8696d8bfd1fc9059e1979f9f79":{"kind":"REWORK","_number":1,"created":"2020-04-02 21:17:30.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/26/717126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/xstatic-jquery.tablesorter","ref":"refs/changes/26/717126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/xstatic-jquery.tablesorter refs/changes/26/717126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/xstatic-jquery.tablesorter refs/changes/26/717126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/xstatic-jquery.tablesorter refs/changes/26/717126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/xstatic-jquery.tablesorter refs/changes/26/717126/1"}}},"commit":{"parents":[{"commit":"4794f24534b68e18c646feabf06f4c9fe267fce3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-jquery.tablesorter/commit/4794f24534b68e18c646feabf06f4c9fe267fce3"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-04-02 21:06:36.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-04-02 21:15:55.000000000","tz":540},"subject":"Prepare for relase using OpenStack release process","message":"Prepare for relase using OpenStack release process\n\nxstatic-jquery.tablesorter has never been released using the OpenStack\nrelease prociess with openstack/releases repository.\nTo sync the release history in openstack/releases and PyPI, we need to\nrelease a new version, so the xstatic version is bumped.\n\nREADME is also updated to satisfy what PyPI expects.\nThis is required to pass the validation in openstack/releases repo.\n\nChange-Id: Ibe7de2ab94752d0fc3a829c93da3e7f9bede8aa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-jquery.tablesorter/commit/09a700b565e3fa8696d8bfd1fc9059e1979f9f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-jquery.tablesorter/commit/09a700b565e3fa8696d8bfd1fc9059e1979f9f79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
