)]}'
{"id":"openstack%2Fxstatic-jquery.tablesorter~787031","triplet_id":"openstack%2Fxstatic-jquery.tablesorter~master~Iab2f29449ba29bd3fc834dab1fef38384880db1a","project":"openstack/xstatic-jquery.tablesorter","branch":"master","hashtags":[],"change_id":"Iab2f29449ba29bd3fc834dab1fef38384880db1a","subject":"Use py3 as the default runtime for tox","status":"ABANDONED","created":"2021-04-20 02:32:59.000000000","updated":"2021-12-16 08:43:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5f939f9effb54c0e302a7850bc51fe83358e0d7d","_number":787031,"virtual_id_number":787031,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-20 02:33:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-1,"date":"2021-04-28 12:04:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 02:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 12:04:12.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"086b81e7a21719e3321f971e065947abe2be5d54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-04-20 02:32:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a2e89a619a0eafd13e78c5b8de6f52023200e76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 02:33:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/68bbaac1e1b84a00a4f66dc43f7a3712 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"0dbe06927a4765a353206d44742303d63eb21912","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-04-28 12:04:12.000000000","message":"Patch Set 1: Code-Review-1\n\nSimilar comment as for https://review.opendev.org/c/openstack/xstatic-rickshaw/+/787024","accounts_in_message":[],"_revision_number":1},{"id":"5f939f9effb54c0e302a7850bc51fe83358e0d7d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-12-16 08:43:19.000000000","message":"Abandoned\n\ndue to long inactivity and no response from the author","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27dbfbe86a18f621c597537f38692ee2c6f2b955","revisions":{"27dbfbe86a18f621c597537f38692ee2c6f2b955":{"kind":"REWORK","_number":1,"created":"2021-04-20 02:32:59.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/31/787031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/xstatic-jquery.tablesorter","ref":"refs/changes/31/787031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/xstatic-jquery.tablesorter refs/changes/31/787031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/xstatic-jquery.tablesorter refs/changes/31/787031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/xstatic-jquery.tablesorter refs/changes/31/787031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/xstatic-jquery.tablesorter refs/changes/31/787031/1"}}},"commit":{"parents":[{"commit":"09a700b565e3fa8696d8bfd1fc9059e1979f9f79","subject":"Prepare for relase using OpenStack release process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-jquery.tablesorter/commit/09a700b565e3fa8696d8bfd1fc9059e1979f9f79"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 02:32:19.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 02:32:53.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: Iab2f29449ba29bd3fc834dab1fef38384880db1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-jquery.tablesorter/commit/27dbfbe86a18f621c597537f38692ee2c6f2b955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-jquery.tablesorter/commit/27dbfbe86a18f621c597537f38692ee2c6f2b955"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
