)]}'
{"id":"openstack%2Fxstatic-json2yaml~787035","triplet_id":"openstack%2Fxstatic-json2yaml~master~I29e3e2a820ec4e43308624b44dbbd39f3ebd803f","project":"openstack/xstatic-json2yaml","branch":"master","hashtags":[],"change_id":"I29e3e2a820ec4e43308624b44dbbd39f3ebd803f","subject":"Use py3 as the default runtime for tox","status":"ABANDONED","created":"2021-04-20 02:37:45.000000000","updated":"2021-12-16 08:43:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e1fb78960449fc854300db4ecdb0b0c42540d07c","_number":787035,"virtual_id_number":787035,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-20 02:37:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-1,"date":"2021-04-28 12:09:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 02:37:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 12:09:32.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"b894ba0b07ac2e7878a3f994f98fb3db33d87d86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-04-20 02:37:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72d8435213976422e1d3e3123a6feecb37882d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 02:37:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/1b718c79119448e895f37cd4d3e42a3b : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"2b11c496482a627fae2aa1c38f7101d9e08fd6f5","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-04-28 12:09:32.000000000","message":"Patch Set 1: Code-Review-1\n\nSimilar comment as for https://review.opendev.org/c/openstack/xstatic-rickshaw/+/787024","accounts_in_message":[],"_revision_number":1},{"id":"e1fb78960449fc854300db4ecdb0b0c42540d07c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-12-16 08:43:52.000000000","message":"Abandoned\n\ndue to long inactivity and no response from the author","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7695b2f5b74b5b7cb37a98a4160db5db62cf0496","revisions":{"7695b2f5b74b5b7cb37a98a4160db5db62cf0496":{"kind":"REWORK","_number":1,"created":"2021-04-20 02:37:45.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/35/787035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/xstatic-json2yaml","ref":"refs/changes/35/787035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/35/787035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/35/787035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/35/787035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/35/787035/1"}}},"commit":{"parents":[{"commit":"6166348bd3d7b0908781f39c8f9c4c981b3f746e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-json2yaml/commit/6166348bd3d7b0908781f39c8f9c4c981b3f746e"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 02:37:08.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 02:37:40.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: I29e3e2a820ec4e43308624b44dbbd39f3ebd803f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-json2yaml/commit/7695b2f5b74b5b7cb37a98a4160db5db62cf0496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-json2yaml/commit/7695b2f5b74b5b7cb37a98a4160db5db62cf0496"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
