)]}'
{"id":"openstack%2Fxstatic-json2yaml~976411","triplet_id":"openstack%2Fxstatic-json2yaml~master~I10626c6b3cfee801bd8a9f1c696ea0fa96257efe","project":"openstack/xstatic-json2yaml","branch":"master","topic":"remove-pkg_resources","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-06 15:53:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I10626c6b3cfee801bd8a9f1c696ea0fa96257efe","subject":"Remove unnecessary setup.cfg file","status":"ABANDONED","created":"2026-02-11 14:34:55.000000000","updated":"2026-05-06 15:53:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"27ef122f387c3b8e50ce930c3efa41df262f14c1","_number":976411,"virtual_id_number":976411,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-11 14:56:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-11 14:42:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"948c65e9ec6288f22ab7a8c81dc4044ae58958a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 14:34:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7cfb79bfc884621805474affdad049aca743472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 14:42:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20f6dc3bd38144fca297afe9aa18609f\n\n- noop https://zuul.opendev.org/t/openstack/build/69277ea770a14b869b64a65022243268 : SUCCESS in 0s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b8c57183f4da41f98a8a14c683333060 : FAILURE in 2m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0805c714b51aadb162c6e36eec2e814cb71d7388","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 14:46:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"62ea3521640581dc364cf2d27018fd5dce6e7b8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 14:56:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3a0cf61b3a04501a130b1068b48aad8\n\n- noop https://zuul.opendev.org/t/openstack/build/f60d091fdddd43569d8e461b4922f638 : SUCCESS in 0s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/13891b547686428f98e62add1dbcb805 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":2},{"id":"27ef122f387c3b8e50ce930c3efa41df262f14c1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-06 15:53:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"483561270b855fcde36e7b8cd64da8262d4e3ef6","revisions":{"4d7df7264f6524f9a3da25b3c0e463df4b55ad58":{"kind":"REWORK","_number":1,"created":"2026-02-11 14:34:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/976411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/xstatic-json2yaml","ref":"refs/changes/11/976411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/11/976411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/11/976411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/11/976411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/11/976411/1"}}},"commit":{"parents":[{"commit":"ccb286ff6d4cb331ee6044724729a7ec2f681a53","subject":"Migrate to native namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-json2yaml/commit/ccb286ff6d4cb331ee6044724729a7ec2f681a53"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 14:25:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 14:25:35.000000000","tz":0},"subject":"Remove unnecessary setup.cfg file","message":"Remove unnecessary setup.cfg file\n\nThis duplicates information from setup.py\n\nChange-Id: I10626c6b3cfee801bd8a9f1c696ea0fa96257efe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-json2yaml/commit/4d7df7264f6524f9a3da25b3c0e463df4b55ad58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-json2yaml/commit/4d7df7264f6524f9a3da25b3c0e463df4b55ad58"}]},"branch":"refs/heads/master"},"483561270b855fcde36e7b8cd64da8262d4e3ef6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-11 14:46:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/976411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/xstatic-json2yaml","ref":"refs/changes/11/976411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/11/976411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/11/976411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/11/976411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/xstatic-json2yaml refs/changes/11/976411/2"}}},"commit":{"parents":[{"commit":"1f6ed70f4be3e464225f89f42944bd3f6201fc85","subject":"Migrate to native namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-json2yaml/commit/1f6ed70f4be3e464225f89f42944bd3f6201fc85"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 14:25:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-11 14:46:21.000000000","tz":0},"subject":"Remove unnecessary setup.cfg file","message":"Remove unnecessary setup.cfg file\n\nThis duplicates information from setup.py\n\nChange-Id: I10626c6b3cfee801bd8a9f1c696ea0fa96257efe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-json2yaml/commit/483561270b855fcde36e7b8cd64da8262d4e3ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-json2yaml/commit/483561270b855fcde36e7b8cd64da8262d4e3ef6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
