)]}'
{"id":"openstack%2Fxstatic-rickshaw~787024","triplet_id":"openstack%2Fxstatic-rickshaw~master~I24d0f68ade8920b52a9552ab55fc5fc478eef61a","project":"openstack/xstatic-rickshaw","branch":"master","hashtags":[],"change_id":"I24d0f68ade8920b52a9552ab55fc5fc478eef61a","subject":"Use py3 as the default runtime for tox","status":"ABANDONED","created":"2021-04-20 02:24:43.000000000","updated":"2021-12-16 08:39:43.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"4d9204252412eb35e6c877954db60a33aa2534bd","_number":787024,"virtual_id_number":787024,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-20 02:24:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":-1,"date":"2021-04-28 12:02:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":-1,"date":"2021-05-05 20:40:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 02:24:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 12:02:35.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2021-05-05 20:40:47.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"c8ff3aeea93596aa3b2f08b5023c3e746f17a69b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-04-20 02:24:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ffb9cb9f36b18d1708c9b83e954a81363f9de41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 02:24:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/e965e3863c7445a9b234d775475b1cd6 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"05d1f139a7c1eb21bd6cb181c8e1e6012ac8f34d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-04-28 12:02:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nCould you please add more info. about why we really need this change.\nAfter your change if you run \u0027tox -e py3\u0027 then you use default python3\non the user platform, not the one requested for the testing.","accounts_in_message":[],"_revision_number":1},{"id":"8b7ba002f5e1dc3155b2b9bc454d724b1eadad8a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-05-05 20:40:47.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Could you please add more info. about why we really need this change.\n\u003e After your change if you run \u0027tox -e py3\u0027 then you use default python3\n\u003e on the user platform, not the one requested for the testing.\n\nIn addition to Vishal\u0027s question, I have another question.\nThis repository does not have UT (from its nature), so both py3 (proposed) and the current (py27 and so on) do not make sense. If you would like to move this patch forward, could you clarify how this change is useful?","accounts_in_message":[],"_revision_number":1},{"id":"4d9204252412eb35e6c877954db60a33aa2534bd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-12-16 08:39:43.000000000","message":"Abandoned\n\ndue to long inactivity and no response from the author.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a3db9b7efa87a1b6338c1ee72104e8b76f45869","revisions":{"6a3db9b7efa87a1b6338c1ee72104e8b76f45869":{"kind":"REWORK","_number":1,"created":"2021-04-20 02:24:43.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/24/787024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/xstatic-rickshaw","ref":"refs/changes/24/787024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/xstatic-rickshaw refs/changes/24/787024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/xstatic-rickshaw refs/changes/24/787024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/xstatic-rickshaw refs/changes/24/787024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/xstatic-rickshaw refs/changes/24/787024/1"}}},"commit":{"parents":[{"commit":"8ceea48e79c5abc864c64012e2bde239afea011c","subject":"Prepare for relase using OpenStack release process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-rickshaw/commit/8ceea48e79c5abc864c64012e2bde239afea011c"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 02:24:07.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 02:24:38.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: I24d0f68ade8920b52a9552ab55fc5fc478eef61a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-rickshaw/commit/6a3db9b7efa87a1b6338c1ee72104e8b76f45869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-rickshaw/commit/6a3db9b7efa87a1b6338c1ee72104e8b76f45869"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
