)]}'
{"id":"openstack%2Fxstatic-spin~787026","triplet_id":"openstack%2Fxstatic-spin~master~I285c2e869966d5dbbe411cb3dbaedfe89a702a83","project":"openstack/xstatic-spin","branch":"master","hashtags":[],"change_id":"I285c2e869966d5dbbe411cb3dbaedfe89a702a83","subject":"Use py3 as the default runtime for tox","status":"ABANDONED","created":"2021-04-20 02:26:58.000000000","updated":"2021-12-16 08:40:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"49d6934c846044d92ad1e68dbd3b67ccf564ac02","_number":787026,"virtual_id_number":787026,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-20 02:27:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-1,"date":"2021-04-28 12:03:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 02:27:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 12:03:46.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"135a05cefd8e93a2eddb4a7c5c5d4e40a07dd572","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-04-20 02:26:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9052a797a35e5b9da2113d1ebf2f3b572cf4294a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 02:27:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/51895494a88e40cb8fa4e5ef395957f4 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"a791c3ab52c39f4d0a18627ff5625d886f4ce58b","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-04-28 12:03:46.000000000","message":"Patch Set 1: Code-Review-1\n\nSimilar comment as for https://review.opendev.org/c/openstack/xstatic-rickshaw/+/787024","accounts_in_message":[],"_revision_number":1},{"id":"49d6934c846044d92ad1e68dbd3b67ccf564ac02","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-12-16 08:40:34.000000000","message":"Abandoned\n\ndue to long inactivity and no response from the author","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea9bb9a8b1cfa304358571a833ab793eeb524d6a","revisions":{"ea9bb9a8b1cfa304358571a833ab793eeb524d6a":{"kind":"REWORK","_number":1,"created":"2021-04-20 02:26:58.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/26/787026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/xstatic-spin","ref":"refs/changes/26/787026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/xstatic-spin refs/changes/26/787026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/xstatic-spin refs/changes/26/787026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/xstatic-spin refs/changes/26/787026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/xstatic-spin refs/changes/26/787026/1"}}},"commit":{"parents":[{"commit":"14237451bad092de9c352d48ef7e0ce1c85d7a58","subject":"Prepare for relase using OpenStack release process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-spin/commit/14237451bad092de9c352d48ef7e0ce1c85d7a58"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 02:26:24.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 02:26:53.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: I285c2e869966d5dbbe411cb3dbaedfe89a702a83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-spin/commit/ea9bb9a8b1cfa304358571a833ab793eeb524d6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-spin/commit/ea9bb9a8b1cfa304358571a833ab793eeb524d6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
