)]}'
{"id":"openstack%2Fxstatic-vis~650842","triplet_id":"openstack%2Fxstatic-vis~master~Ice2d3426b0e2b44971cd814e367a247ffba29a2c","project":"openstack/xstatic-vis","branch":"master","hashtags":[],"change_id":"Ice2d3426b0e2b44971cd814e367a247ffba29a2c","subject":"Retire project","status":"MERGED","created":"2019-04-08 10:33:59.000000000","updated":"2019-04-24 15:59:10.000000000","submitted":"2019-04-24 15:59:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"650842-1556121550658-81340386","meta_rev_id":"5340242cdad830e7638fe05ae6202b28662a5880","_number":650842,"virtual_id_number":650842,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-04-24 15:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2019-04-24 15:42:30.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2019-04-24 15:58:55.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2019-04-24 15:58:55.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 15:42:30.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2019-04-24 15:58:55.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-04-24 15:59:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f68889779d09e3d8cc090464fcc0c7ed78d2b797","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-08 10:33:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"295dae7e31833fa1f79443e599eb29857247ae14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 10:34:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop noop : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"2b4c50547cac9f6f3643163e1f5dbf92f2000505","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-24 15:09:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"48493d87d179cd4ef719deedb9af8a5dea859f02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:09:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- noop noop : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"4011465b397f68c73403e456f121a3d0e798df77","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-04-24 15:42:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2380b3249e1f0d21fb615a4569a7b08befe1a695","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-04-24 15:58:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0abe9193eb32f3acf5423997bdd4ab4e1be3b94d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:59:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"336e34ee73483c10dc972f26240d6395c5f954c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:59:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop noop : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"5340242cdad830e7638fe05ae6202b28662a5880","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:59:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a6d942e8cc084b3b163738a013dc0e84b3d7e2a6","revisions":{"1c5acb689a3c10ae5f281e84680ff2de89dfae95":{"kind":"REWORK","_number":1,"created":"2019-04-08 10:33:59.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/42/650842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/xstatic-vis","ref":"refs/changes/42/650842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/xstatic-vis refs/changes/42/650842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/xstatic-vis refs/changes/42/650842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/xstatic-vis refs/changes/42/650842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/xstatic-vis refs/changes/42/650842/1"}}},"commit":{"parents":[{"commit":"bfb51c0009f64f31e66fc26e3655b00449fa22c2","subject":"fix tox python3 overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-vis/commit/bfb51c0009f64f31e66fc26e3655b00449fa22c2"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-08 10:33:47.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-08 10:33:47.000000000","tz":180},"subject":"Retire project","message":"Retire project\n\nThis project is not used by any other one we are not going\nto do it in the future.\n\nChange-Id: Ice2d3426b0e2b44971cd814e367a247ffba29a2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-vis/commit/1c5acb689a3c10ae5f281e84680ff2de89dfae95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-vis/commit/1c5acb689a3c10ae5f281e84680ff2de89dfae95"}]},"branch":"refs/heads/master"},"a6d942e8cc084b3b163738a013dc0e84b3d7e2a6":{"kind":"REWORK","_number":2,"created":"2019-04-24 15:09:44.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/42/650842/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/xstatic-vis","ref":"refs/changes/42/650842/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/xstatic-vis refs/changes/42/650842/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/xstatic-vis refs/changes/42/650842/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/xstatic-vis refs/changes/42/650842/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/xstatic-vis refs/changes/42/650842/2"}}},"commit":{"parents":[{"commit":"d285037a57eb585cf1380e4456497629c2e4c208","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-vis/commit/d285037a57eb585cf1380e4456497629c2e4c208"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-08 10:33:47.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-24 15:09:36.000000000","tz":180},"subject":"Retire project","message":"Retire project\n\nThis project is not used by any other one we are not going\nto do it in the future.\n\nChange-Id: Ice2d3426b0e2b44971cd814e367a247ffba29a2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-vis/commit/a6d942e8cc084b3b163738a013dc0e84b3d7e2a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/xstatic-vis/commit/a6d942e8cc084b3b163738a013dc0e84b3d7e2a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
