)]}'
{"id":"openstack%2Fyaql~912686","triplet_id":"openstack%2Fyaql~master~I708168743594f4a40f9e8812062b1d9588fe40fd","project":"openstack/yaql","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-13 08:20:13.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I708168743594f4a40f9e8812062b1d9588fe40fd","subject":"Publish release notes","status":"MERGED","created":"2024-03-13 01:36:50.000000000","updated":"2024-03-13 08:43:49.000000000","submitted":"2024-03-13 08:41:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912686","meta_rev_id":"2b4349841fce1c28f3e818efd51bea15243bb55e","_number":912686,"virtual_id_number":912686,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-13 08:41:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-03-13 08:20:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-03-13 08:20:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-13 01:49:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb5d6350c9d0bc5afd33ad253c761e7a18d43faa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-13 01:36:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0eb5149450e2b9ef2c92b4c884dcc4c48a3506d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 01:49:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34d3357a1c61454383b0face4998ea3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a037397dce04805bccbecf134a520ef : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2afb3cf4c7774be3949985dc7147ece7 : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1822d22d4584730965af12ebe1a311b : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/155f31e2718a4462a842a3de69fbad7f : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fa10ed98bc6483e94287c1aa880afa7 : FAILURE in 2m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c5e1bcb44a4d3b07ca8522d54b9fe8f4d1617399","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-13 02:58:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8e5156945fbb028ca75f5ffd27dd66cfb69dc5e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-13 03:00:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a3c88a1ea13061b2ffb6e4701fbcccd75619f381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 03:28:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd5a6de8138a41f287972f0fbb808e81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd56297cb1504c3a9e554df4958cc258 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45995e84b1654bff8e9f711e17e546b9 : SUCCESS in 2m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b11583e2ee1d4431a2878982563ec402 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e24efc6e1b3e46c89a956eb5bd902ae6 : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98561b63bce042e9bb8570ee9038acf4 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":3},{"id":"20ef5b7dbdf33f6f3386a16c880728bf8f4fcc28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-13 08:06:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d65fd59795eb63346f50a4e31d2daf3e72a697f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-13 08:06:57.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"dba0166ec0209bc55fb25068555d1afe84008653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 08:15:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48bac8b4f8ef41ad8b8ba7489e20ae7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35662851c69649b7a4123d9c92db7ae3 : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54aa66d932cd40f2b1cb52051858ad6d : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd20804fb1b943c78515f7700231dfd5 : SUCCESS in 2m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f75a64baec434c199700e9d51c71af7d : SUCCESS in 2m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ee1dbb3f2b54634bb335f4b89834827 : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":5},{"id":"850b8b6d4e208ff7db2083d59d812f82c9913a56","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-13 08:20:13.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"df0a95cdb6aa1d6d28a94c37de9d5b5d9d99cba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 08:20:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"da79f72b5909fe168edb43397fdc1a527168a77b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 08:41:22.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47fea24d947f4a5eadd1a71a6e59c12d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20405182cf6c4be1abcda88039c95e25 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b5337058249414d9afc1fa3db53c2e4 : SUCCESS in 2m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/965e4c958b2044a293a2ab91190cfcd8 : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a9896ae2ec5426b8dbeaaa156848578 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2cb69321ec349efb2fc4930f113a52f : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":5},{"id":"7adc0bdda276f5f55ceba5840f38ffbe61fdd92e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 08:41:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2b4349841fce1c28f3e818efd51bea15243bb55e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 08:43:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57bc7e0deeb5430886d5c1872635207f\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c5bda1dfb5a4f38b0f8efb26fb5bb64 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2a159b85ae02f74645b22822e40fcbccf495493f","revisions":{"7737db12dca3e9b7caebc388ff643dedf4f2a413":{"kind":"REWORK","_number":1,"created":"2024-03-13 01:36:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/912686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/yaql","ref":"refs/changes/86/912686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/yaql refs/changes/86/912686/1"}}},"commit":{"parents":[{"commit":"0d63eb1346c829c0ce818dd2e00590af6330a0f3","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/0d63eb1346c829c0ce818dd2e00590af6330a0f3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 01:33:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 01:33:44.000000000","tz":540},"subject":"Publish release notes","message":"Publish release notes\n\nChange-Id: I708168743594f4a40f9e8812062b1d9588fe40fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/7737db12dca3e9b7caebc388ff643dedf4f2a413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/7737db12dca3e9b7caebc388ff643dedf4f2a413"}]},"branch":"refs/heads/master"},"3bd938a3ae28bc377d2acb5ec862ad9148f1e4e2":{"kind":"REWORK","_number":2,"created":"2024-03-13 02:58:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/912686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/yaql","ref":"refs/changes/86/912686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/yaql refs/changes/86/912686/2"}}},"commit":{"parents":[{"commit":"0d63eb1346c829c0ce818dd2e00590af6330a0f3","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/0d63eb1346c829c0ce818dd2e00590af6330a0f3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 01:33:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 02:58:20.000000000","tz":540},"subject":"Publish release notes","message":"Publish release notes\n\nChange-Id: I708168743594f4a40f9e8812062b1d9588fe40fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/3bd938a3ae28bc377d2acb5ec862ad9148f1e4e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/3bd938a3ae28bc377d2acb5ec862ad9148f1e4e2"}]},"branch":"refs/heads/master"},"0d4fc679e3b2b83c3ed5e879d5dda3ca0e8973b6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-13 03:00:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/912686/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/yaql","ref":"refs/changes/86/912686/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/yaql refs/changes/86/912686/3"}}},"commit":{"parents":[{"commit":"22ff7ab75dde05fd2b2c5ecce267cb4087d49d9a","subject":"Add release note about Python 3.6 support removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/22ff7ab75dde05fd2b2c5ecce267cb4087d49d9a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 01:33:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 03:00:28.000000000","tz":540},"subject":"Publish release notes","message":"Publish release notes\n\nChange-Id: I708168743594f4a40f9e8812062b1d9588fe40fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/0d4fc679e3b2b83c3ed5e879d5dda3ca0e8973b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/0d4fc679e3b2b83c3ed5e879d5dda3ca0e8973b6"}]},"branch":"refs/heads/master"},"9fff158ebc4f08e644c52a4fbe5a068e275b359b":{"kind":"REWORK","_number":4,"created":"2024-03-13 08:06:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/912686/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/yaql","ref":"refs/changes/86/912686/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/yaql refs/changes/86/912686/4"}}},"commit":{"parents":[{"commit":"22ff7ab75dde05fd2b2c5ecce267cb4087d49d9a","subject":"Add release note about Python 3.6 support removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/22ff7ab75dde05fd2b2c5ecce267cb4087d49d9a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 01:33:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 08:05:35.000000000","tz":540},"subject":"Publish release notes","message":"Publish release notes\n\nChange-Id: I708168743594f4a40f9e8812062b1d9588fe40fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/9fff158ebc4f08e644c52a4fbe5a068e275b359b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/9fff158ebc4f08e644c52a4fbe5a068e275b359b"}]},"branch":"refs/heads/master"},"2a159b85ae02f74645b22822e40fcbccf495493f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-03-13 08:06:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/912686/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/yaql","ref":"refs/changes/86/912686/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/yaql refs/changes/86/912686/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/yaql refs/changes/86/912686/5"}}},"commit":{"parents":[{"commit":"0d63eb1346c829c0ce818dd2e00590af6330a0f3","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/0d63eb1346c829c0ce818dd2e00590af6330a0f3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 01:33:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 08:06:47.000000000","tz":540},"subject":"Publish release notes","message":"Publish release notes\n\nChange-Id: I708168743594f4a40f9e8812062b1d9588fe40fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/2a159b85ae02f74645b22822e40fcbccf495493f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/yaql/commit/2a159b85ae02f74645b22822e40fcbccf495493f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
