)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~1000188","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~Icdbc031a3e550586d1db1aad63daf498c1f73d50","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"release-health-check","attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-08-08 10:30:48.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Icdbc031a3e550586d1db1aad63daf498c1f73d50","subject":"DNM: gate health test","status":"NEW","created":"2026-08-08 09:42:59.000000000","updated":"2026-08-08 10:30:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f98c5b2ef5203362464f22e426e19a4aa010d26","_number":1000188,"virtual_id_number":1000188,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-08 10:30:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-08 10:30:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1590bb936ceab7b18cf5f7793b525b3696bbff30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-08 09:42:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f98c5b2ef5203362464f22e426e19a4aa010d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 10:30:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d4e4a93f9684459b6c93bec077b0d2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c258d15330f4fd1b808886bccf16eec : SUCCESS in 2m 48s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/52f1b029c0204abb833e28d477b0fc3a : SUCCESS in 18m 16s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/b384b05c48d4478a9a26ecaaa84473ba : SUCCESS in 30m 38s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/13f5cded506548a8bbb91e80f6e6cae5 : SUCCESS in 18m 26s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/43b3ae59f7cc4ad383da27ab4b010373 : FAILURE in 32m 38s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/27d738e1db4c4e73828179eb973f4c31 : SUCCESS in 31m 31s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/987383f9bfa24da4a3cf7f3223bf7509 : SUCCESS in 19m 30s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/0736093e38e04f2084ad58d39ff90e91 : SUCCESS in 29m 17s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/31673272ba4548a3bc5e1205478e9925 : FAILURE in 24m 28s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/097dd0e8b93e4bdeaaeb85e5161323af : SUCCESS in 22m 29s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/d69143cf9f364724bc27008aafc5c907 : SUCCESS in 22m 16s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/81f225374fea4cb18caee49362f28c41 : SUCCESS in 30m 47s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/3e7c83fb38224fa1bfc31c4b4bc8fbc4 : SUCCESS in 33m 07s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/d9a9e8ba1cda4b56afeb8419d0c98b29 : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9773ac20a798ba116e52c3d91e2b2947fd00614","revisions":{"c9773ac20a798ba116e52c3d91e2b2947fd00614":{"kind":"REWORK","_number":1,"created":"2026-08-08 09:42:59.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/88/1000188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/88/1000188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/88/1000188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/88/1000188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/88/1000188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/88/1000188/1"}}},"commit":{"parents":[{"commit":"b275825c0b9e7cbce9a1a160586dd6d158ffe8b6","subject":"Drop python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/b275825c0b9e7cbce9a1a160586dd6d158ffe8b6"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-08 09:42:49.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-08 09:42:58.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: Icdbc031a3e550586d1db1aad63daf498c1f73d50\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/repository_test_generator.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/c9773ac20a798ba116e52c3d91e2b2947fd00614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/c9773ac20a798ba116e52c3d91e2b2947fd00614"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
