)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~1005307","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~Iabe7b9b247e7a83e5f8569698bd63c4b8341c014","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"2026.2-job","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-11 21:52:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Iabe7b9b247e7a83e5f8569698bd63c4b8341c014","subject":"Add stable/2026.2 job","status":"MERGED","created":"2026-09-11 17:02:59.000000000","updated":"2026-09-11 23:17:41.000000000","submitted":"2026-09-11 23:17:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005307-2026.2-job","meta_rev_id":"9ffdabf7a0feca6eef3bbf948b4254190d5ef462","_number":1005307,"virtual_id_number":1005307,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 23:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-11 17:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-11 17:46:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 17:46:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ceefd868c467da3c072ac3f02a8c8e5267a0d98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-11 17:02:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0530415e0f95d49f8693911b1ce3594cabedf7db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 17:46:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abc8472682ef466fbfc22d2047cbfc4d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3664eef02fb942ffa0360ea6dc61710a : SUCCESS in 2m 29s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/625a1c28fa204f3fbedfa41ca9267e16 : SUCCESS in 21m 52s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/37abb320dd2d4ecf907aa06a61f5b6b0 : SUCCESS in 21m 39s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/889877ec2904417ebc89e463f4c19766 : SUCCESS in 21m 52s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/38c6b81df7b849f5b8b4a9e16b5a21ce : SUCCESS in 21m 43s\n- zaqar-tempest-plugin-mongodb-2026-2 https://zuul.opendev.org/t/openstack/build/ad7ed09be2844beab3cfe12f3d3d0924 : SUCCESS in 22m 05s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/cf85ca40cd6a4b7aaeb323098bb9b6b6 : SUCCESS in 35m 50s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/319be97a7944402994b892e3daf066f8 : SUCCESS in 30m 48s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/003b5191db4d4ca5989988ecdef1fb9f : SUCCESS in 31m 33s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/fe8ede9e44b748feb3c47e8122b55478 : SUCCESS in 30m 28s\n- zaqar-tempest-plugin-redis-2026-2 https://zuul.opendev.org/t/openstack/build/dff8ccfada21429bbf151fe9dd0ddf4f : SUCCESS in 34m 27s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/1ca8005574f745d1941725861fc415db : SUCCESS in 22m 40s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/98f88d198de9403ea17a4c809dd98339 : SUCCESS in 37m 02s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/144a0d85fa7242789371a25646f54158 : FAILURE in 23m 12s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/752165cb214c46a08a386d5886934a2f : SUCCESS in 25m 56s\n- zaqar-tempest-plugin-swift-2026-2 https://zuul.opendev.org/t/openstack/build/419017b3bae9404ba3cb5d3f729aa64b : SUCCESS in 23m 22s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/464795098d094b5e97c8b48bb62fc498 : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"406724b50429e5d47c2e2a5ce3988ecb1c678f77","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-11 17:46:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e655dcdc4b491e0e7928c2ca5feca2d3e8e3b01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:35:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cb8189b7a424f1da3d0fee1a2088060\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4076a8f353ce42579b8f7250e64c9daf : SUCCESS in 1m 45s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/f332afba34ca42b0acf1ea8a5a942a56 : FAILURE in 16m 57s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/9d93bec783e6407d80558fdd7d47048d : SUCCESS in 18m 14s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/1bfc723c84184106bca04413c996935c : SUCCESS in 17m 10s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/0fa465bd70e04a1fa4b6d5fadc4a8794 : SUCCESS in 32m 01s\n- zaqar-tempest-plugin-mongodb-2026-2 https://zuul.opendev.org/t/openstack/build/8ce10965937b4ae198b2748825b6d0f5 : SUCCESS in 26m 36s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/c19ae297a6f04a76b76262384fe5cc36 : SUCCESS in 33m 06s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/cd14d43a6be54bd0bcb0cb3289e93dcc : SUCCESS in 18m 42s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/a77845bbaa894645ae21472b0aeeee0e : SUCCESS in 15m 09s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/52e3e2d8c7064ea89b89bb606dbe6ab6 : SUCCESS in 16m 57s\n- zaqar-tempest-plugin-redis-2026-2 https://zuul.opendev.org/t/openstack/build/4903e49ab04242699bbced4eca3018ba : SUCCESS in 32m 45s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/89c4d825a0384994a77203aee552858d : SUCCESS in 32m 12s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/886cca42ca0540d7b79cec07aa0dd0f1 : SUCCESS in 18m 14s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/4cad29b485604f3287e82748d6b8c9a6 : SUCCESS in 17m 27s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/12ac7f539ffe4ff180e9ee6e420edfe2 : SUCCESS in 35m 29s\n- zaqar-tempest-plugin-swift-2026-2 https://zuul.opendev.org/t/openstack/build/b320012f92b84a29b88a620e53ab7c2a : SUCCESS in 32m 56s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/c43fc04b10fe4f109ed8fa15e90aeacd : SUCCESS in 25m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3166afac0b76e383d73cf35ac7aa1f6e6b151a34","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-11 21:52:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4231612c47555aa9299ee6dfc3db7b2d967c550b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 22:35:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50a2c82f06c7443e93bbc8787f44cdfa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af66a62c117d4b49a3476136077554b5 : SUCCESS in 2m 43s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/c63eb59b65644570997fa26de379fadb : SUCCESS in 34m 55s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/cd95f48057024afa812baef09638ed02 : SUCCESS in 31m 51s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/fe12e021b93f4a72aca7b1d1843de2a2 : SUCCESS in 34m 09s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/38185dd364d4423f9e7fa7182a39a90b : SUCCESS in 32m 39s\n- zaqar-tempest-plugin-mongodb-2026-2 https://zuul.opendev.org/t/openstack/build/d239952a498a4c59b20b39d345ec1c7a : SUCCESS in 15m 56s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/bf14b9dc2367481db43b82e75d99bd3a : SUCCESS in 33m 32s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/99dfe2e7a70c47b9b9a21df2b6c41a6a : SUCCESS in 30m 52s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/85246e53fd2144f4b196fd6714ab3b6e : SUCCESS in 29m 53s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/348f0a57c617498aae89c216b1dc8a31 : SUCCESS in 34m 44s\n- zaqar-tempest-plugin-redis-2026-2 https://zuul.opendev.org/t/openstack/build/c58af43a12dc4e0fa749e246ba2a6e6e : SUCCESS in 17m 06s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/822ec54071574354a490bdfd3096a7cf : SUCCESS in 33m 03s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/704910aae2644b6bbd51d9ebfe0b088e : SUCCESS in 37m 07s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/5201d47323f5467eb7b27841af40d149 : SUCCESS in 16m 40s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/71315339250740db9279c09d799e0818 : SUCCESS in 34m 48s\n- zaqar-tempest-plugin-swift-2026-2 https://zuul.opendev.org/t/openstack/build/d3a4a04eaa444113b9985d48aed860e1 : SUCCESS in 32m 20s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/81b3735d91924cdcbc4cab6cc0ab3037 : SUCCESS in 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f4fd451ffc7454acc2f0a6894e5a0ff99a34b9f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 22:36:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee7a99af6cf6a3df208bd21e1413b96cbe49f174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 23:17:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6136d131db9547bab5761bbd06f3df9e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12063813eca34ee3869001ed2408acca : SUCCESS in 2m 49s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/7fb7e0edf57e4b0386984506f8bfb7d3 : SUCCESS in 17m 12s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/dec4360377494840a4641c3ef45b1e2d : SUCCESS in 17m 29s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/4b97a142bee14a5c85342fa702255f21 : SUCCESS in 36m 30s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/daa63aafad5f4ca28d8823e065b00295 : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9ffdabf7a0feca6eef3bbf948b4254190d5ef462","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 23:17:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4dbe63018c0b54a12e611fb432d9b31fb0f4b54b","revisions":{"4dbe63018c0b54a12e611fb432d9b31fb0f4b54b":{"kind":"REWORK","_number":1,"created":"2026-09-11 17:02:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/1005307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/07/1005307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/07/1005307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/07/1005307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/07/1005307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/07/1005307/1"}}},"commit":{"parents":[{"commit":"b275825c0b9e7cbce9a1a160586dd6d158ffe8b6","subject":"Drop python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/b275825c0b9e7cbce9a1a160586dd6d158ffe8b6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:02:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:02:09.000000000","tz":540},"subject":"Add stable/2026.2 job","message":"Add stable/2026.2 job\n\nThe branch was created so we should start testing it.\n\nChange-Id: Iabe7b9b247e7a83e5f8569698bd63c4b8341c014\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/4dbe63018c0b54a12e611fb432d9b31fb0f4b54b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/4dbe63018c0b54a12e611fb432d9b31fb0f4b54b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
