)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~1007517","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~Ica4f25e88f80d965cfc0301f0e633e3dc7a39835","project":"openstack/zaqar-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-28 11:51:21.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ica4f25e88f80d965cfc0301f0e633e3dc7a39835","subject":"Skip tempest tests for benchmark tool update","status":"MERGED","created":"2026-09-25 16:25:03.000000000","updated":"2026-09-28 13:21:11.000000000","submitted":"2026-09-28 13:21:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007517","meta_rev_id":"9997340e90b2b978f5ffff7d0bb755016de89d14","_number":1007517,"virtual_id_number":1007517,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-28 13:21:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-26 02:22:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-26 02:22:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 18:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6fe6b55bcef97ea6e20390b5a54aba2abc5f2b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 16:25:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2dfc18c73cf53e2d61788e24ea09a012899fb7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 17:11:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3ccb032e91682d399d857cc79e20acacab96a6eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 18:03:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15172ea20e9b46f1a6f407885080468b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1321a52a69345fe8a79c67869b8714c : SUCCESS in 2m 46s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/9b9aaeb97fe0411593f2e3250eb90996 : SUCCESS in 20m 00s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/4bb2cf3dfc5d481a84776fd363057845 : SUCCESS in 34m 44s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/69ad6641fc0e49139594a67ed2acb0a7 : SUCCESS in 32m 03s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/97e864498718424a8d0ae52882219412 : SUCCESS in 20m 19s\n- zaqar-tempest-plugin-mongodb-2026-2 https://zuul.opendev.org/t/openstack/build/577a1d379b4049cf8429cb259c3456b1 : SUCCESS in 33m 36s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/081b3320d0f8448b85273c5e019f07bd : SUCCESS in 20m 03s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/668b42df1082481fbdb571f075a28a9e : SUCCESS in 20m 02s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/7acb95c5672042b7817007a3781b5dc8 : SUCCESS in 20m 20s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/976a879d97b34fdf805781905406f41a : SUCCESS in 19m 58s\n- zaqar-tempest-plugin-redis-2026-2 https://zuul.opendev.org/t/openstack/build/43cc5af505d1407394577e796441e76c : SUCCESS in 19m 27s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/db66051985f14ec5a398f98f1fcdfdad : SUCCESS in 21m 12s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/d1195d2b1fee480f83f636e4f9e88af6 : SUCCESS in 21m 00s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/8035182bec324b46865c0154b70e243f : SUCCESS in 20m 52s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/670c6c4fa9124ed6bfbc99670859d47e : SUCCESS in 38m 24s\n- zaqar-tempest-plugin-swift-2026-2 https://zuul.opendev.org/t/openstack/build/ff35400851884023b592b3b82e448e29 : SUCCESS in 21m 00s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/3c0418d2ad004f1e9be09c9f25f42856 : SUCCESS in 21m 37s","accounts_in_message":[],"_revision_number":2},{"id":"44e6893b336229157627d09d8a64e77436e8f10d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-26 02:22:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7c4fdc0d8406aced39ce4cdae10b1e92333271ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 02:58:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fff72bad50fe6320ec4dc44dba9534ccce970191","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 03:37:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9741d33893924618a7c134b73eba01f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70dc114679ba461e9cd9f6e57c9f6f39 : SUCCESS in 3m 26s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/fc76231a18e54e77bd907fbb2965559a : FAILURE in 33m 31s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/c355374f015546c1b331a5b4fbbb7774 : SUCCESS in 15m 59s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/224c83c45d5544069c04f95681b4ebcc : SUCCESS in 19m 32s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/5de7a2ad8fff414f9ab8775f313dba03 : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":2},{"id":"2e7936cadc92a9b7582675af192423f048cd06dd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-26 14:20:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b5ca064684ca835c0b04944c69cee4a2f9e84ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 15:01:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac1258c4bf4a4dc89b57cc2ef2cc0f6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cbd3a1777994f0b9edefa5d69c1bdde : SUCCESS in 1m 45s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/0bbd285c7c74443ab4dbcc6038acfeb6 : SUCCESS in 33m 30s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/a0e437722b5d47a08dd9f1369b568b90 : SUCCESS in 32m 31s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/9814db98b4d7416390122d109591d258 : SUCCESS in 15m 09s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/280a8ca6c9674b569e00ba8d5daa855f : SUCCESS in 30m 27s\n- zaqar-tempest-plugin-mongodb-2026-2 https://zuul.opendev.org/t/openstack/build/088eca89ff904563b0fe53609f33ba9e : FAILURE in 29m 18s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/09448eacd4e247eca7ea7cdb4ed36aed : SUCCESS in 29m 48s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/71ce59dc4c674452bf9ecf54081b9dcf : SUCCESS in 30m 23s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/6343d26f8928422a92e380342fcf5149 : SUCCESS in 34m 25s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/a41a66f1e0694c6bba71ff47a442d64c : SUCCESS in 33m 03s\n- zaqar-tempest-plugin-redis-2026-2 https://zuul.opendev.org/t/openstack/build/21a1b3cade334e1baafc5ba0f5fe1de5 : FAILURE in 20m 28s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/0761ae719e1c495d8e08bc726fce23ec : SUCCESS in 31m 23s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/7f1a90e7f88f4a8caf67fddb37e88fa9 : SUCCESS in 18m 07s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/2d820263a9624a628b60f25465a67bfd : SUCCESS in 18m 07s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/2fe57fa0b81b47eda8af3814b812d9db : FAILURE in 19m 16s\n- zaqar-tempest-plugin-swift-2026-2 https://zuul.opendev.org/t/openstack/build/29c430805a1e4ddaa4f0fcaec44de31d : SUCCESS in 23m 59s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/60e2b02279d34d63aa3c70f54eb5bad6 : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":2},{"id":"141dec834732de312d368c7ced772894e0dbd6b9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-26 15:49:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0fcbe473bbab53a8a65df3c854f8b9b281e58e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 16:27:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c93e858b06ed40a294a89ab1add131b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c10b4b19734189a2cbf27ff91d73de : SUCCESS in 3m 39s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/b6fa1403c8ba4d0a9b43124c927ccd24 : SUCCESS in 30m 54s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/a8620ab941d249b882617f94d89d5ea9 : SUCCESS in 16m 41s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/46475b149ef24fc6b3d0bcdb1707df15 : SUCCESS in 30m 11s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/d7bd691afc844fe78410d3d3491e2cd1 : SUCCESS in 31m 16s\n- zaqar-tempest-plugin-mongodb-2026-2 https://zuul.opendev.org/t/openstack/build/e61d5e1beb4940efa764ded0d44e99b8 : SUCCESS in 16m 15s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/9852767e181442cf85f8bb1f360dc32a : SUCCESS in 19m 40s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/594e49ecd72049698e5045c53e803777 : FAILURE in 29m 28s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/4791d4ac665e4786ac5a3cf88fb455d7 : SUCCESS in 20m 10s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/19c2956b52294892a5180ac27325467a : SUCCESS in 18m 18s\n- zaqar-tempest-plugin-redis-2026-2 https://zuul.opendev.org/t/openstack/build/335811f1d16740a2a0737af96db49b2a : SUCCESS in 28m 59s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/ed1751d480ed4498807473d2ca5d4e47 : SUCCESS in 16m 45s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/e773f8f201b64b2eb19ad7a0656b47e6 : SUCCESS in 32m 20s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/a042d52f629a4565bdd3b50f2dee8581 : SUCCESS in 32m 40s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/7654239d191c482990ed0f438de4d43f : SUCCESS in 18m 25s\n- zaqar-tempest-plugin-swift-2026-2 https://zuul.opendev.org/t/openstack/build/d8b4bce623d14b3587240644b96a0720 : SUCCESS in 18m 45s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/b09434ddfb4d489bb1240711b4c0fbaa : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":2},{"id":"5e45fb0bdf4d21461a6124144e1d7c8dfaf53b53","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-28 11:51:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee693a788924842594a116d9b653030602c6ddd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 12:35:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22e7537a0de242469d4fa6c0ec267a72\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35cce8ad47df46ea972dfd5841976845 : SUCCESS in 1m 57s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/da6517399ea54f6582944251fb0acce6 : SUCCESS in 18m 49s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/30a7ef9e4ed54715b8b0ac0698cd9059 : SUCCESS in 19m 21s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/308e8e9143184f71bef06c1c6e80a6dc : SUCCESS in 16m 33s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/05c922dff5a846e0aaf4af2fce061dd3 : SUCCESS in 17m 14s\n- zaqar-tempest-plugin-mongodb-2026-2 https://zuul.opendev.org/t/openstack/build/8f47f8ebe4a649a481e12ef4b70099c0 : SUCCESS in 35m 19s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/56c0aac20d7943f7b844f37487ac6995 : SUCCESS in 35m 33s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/8269ee19dadc443bb1d2498183c4e57b : SUCCESS in 16m 24s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/f8500ee176fd4f93bdb74a0f5cd9429d : SUCCESS in 17m 39s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/b14f4707357643ecb04698e049c69142 : SUCCESS in 33m 08s\n- zaqar-tempest-plugin-redis-2026-2 https://zuul.opendev.org/t/openstack/build/6c016180c32f4a538b68152a049bd5d2 : SUCCESS in 21m 49s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/d73b1ae689034f218377f003884fd4fc : SUCCESS in 31m 13s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/6e46a9a120a44463ba52689d5f96fc30 : SUCCESS in 33m 51s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/dae27458563e47e3a2c345820bb92ab7 : SUCCESS in 32m 20s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/9f66a63f704b49448a2a1a97226ef9bd : SUCCESS in 36m 47s\n- zaqar-tempest-plugin-swift-2026-2 https://zuul.opendev.org/t/openstack/build/bdeef472305344679d924bb06966a7a3 : SUCCESS in 17m 19s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/7b571f29fac040ef964d7d4df79a5cd3 : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"908ffca309460e3f5130395f3e91124cccbff196","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 12:35:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"953931253d58154c4357673ea162f1ecc1d56f63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 13:21:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e542b7daed44ee58498985c220b2d9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71bdf7bd6dc14dd38eba8bc6f424e717 : SUCCESS in 2m 29s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/d42ada9481494084b5c60829fe79420b : SUCCESS in 39m 21s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/2718fa4e27314fec83fa5030b4e7b361 : SUCCESS in 37m 42s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/0a81e2915bb6497e8553a2e17dab0a7c : SUCCESS in 19m 12s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/a0481806419346eab469d35c91c13fec : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":2},{"id":"9997340e90b2b978f5ffff7d0bb755016de89d14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 13:21:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59181609be36a966c798253714921e08848b9e51","revisions":{"80ba83bf1ab6f74abb4d2c9996d3450f19cd1b4a":{"kind":"REWORK","_number":1,"created":"2026-09-25 16:25:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/1007517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/17/1007517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/17/1007517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/17/1007517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/17/1007517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/17/1007517/1"}}},"commit":{"parents":[{"commit":"31e733e084a4705522d7787fb454ce58a9479951","subject":"Use base plugin interface to load additional client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/31e733e084a4705522d7787fb454ce58a9479951"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 16:24:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 16:24:39.000000000","tz":540},"subject":"Skip tempest tests for benchmark tool update","message":"Skip tempest tests for benchmark tool update\n\n.. because the tool is not used in these tests.\n\nChange-Id: Ica4f25e88f80d965cfc0301f0e633e3dc7a39835\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/80ba83bf1ab6f74abb4d2c9996d3450f19cd1b4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/80ba83bf1ab6f74abb4d2c9996d3450f19cd1b4a"}]},"branch":"refs/heads/master"},"59181609be36a966c798253714921e08848b9e51":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-25 17:11:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/1007517/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/17/1007517/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/17/1007517/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/17/1007517/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/17/1007517/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/17/1007517/2"}}},"commit":{"parents":[{"commit":"4dbe63018c0b54a12e611fb432d9b31fb0f4b54b","subject":"Add stable/2026.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/4dbe63018c0b54a12e611fb432d9b31fb0f4b54b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 16:24:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 17:11:52.000000000","tz":0},"subject":"Skip tempest tests for benchmark tool update","message":"Skip tempest tests for benchmark tool update\n\n.. because the tool is not used in these tests.\n\nChange-Id: Ica4f25e88f80d965cfc0301f0e633e3dc7a39835\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/59181609be36a966c798253714921e08848b9e51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/59181609be36a966c798253714921e08848b9e51"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"31e733e084a4705522d7787fb454ce58a9479951","ours":"80ba83bf1ab6f74abb4d2c9996d3450f19cd1b4a","theirs":"4dbe63018c0b54a12e611fb432d9b31fb0f4b54b","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
