)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~536337","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~I26b839a61ea68202ac15e0c0ce00a141f1753e02","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"openstack/requirements","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-26 13:26:16.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I26b839a61ea68202ac15e0c0ce00a141f1753e02","subject":"Updated from global requirements","status":"MERGED","created":"2018-01-22 11:21:54.000000000","updated":"2024-10-26 14:31:31.000000000","submitted":"2024-10-26 14:31:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"536337-openstack/requirements","meta_rev_id":"9a4d35e8abb7b9e0dce66bb754772d9ade76affe","_number":536337,"virtual_id_number":536337,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-26 14:31:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-10-24 19:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-10-24 19:19:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-15 09:53:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-24 19:19:48.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"1889cffb113906921d29c3d8277c996c59223ceb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-01-22 11:21:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"626117ca79539ee69be71d336797925f89404c6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-22 11:32:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/37/536337/1/check/requirements-check/703c3cf/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/37/536337/1/check/openstack-tox-pep8/6d389ea/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0f39019693c3c5a731779475e2f526294eb9e851","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-01-24 01:42:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d045e7c24a824c7e5acad61246fdf1fe98cadd79","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-01-24 02:31:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"29035187d33b6f2e0cda113c1ccc97555ccd4635","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-24 03:53:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/37/536337/3/check/requirements-check/b4bccd1/ : FAILURE in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/37/536337/3/check/openstack-tox-pep8/8e32ee7/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"710a88742bcdcb4593f2b645a223f263c782e8a9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-01-27 18:51:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f8638e596d9a7298b26c766cd4a5de1c983b3f18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-27 19:13:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/37/536337/4/check/requirements-check/c90fce6/ : SUCCESS in 3m 12s\n- openstack-tox-pep8 http://logs.openstack.org/37/536337/4/check/openstack-tox-pep8/9e5c515/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":4},{"id":"085dc8245b00ea4aba6210dc8a14b0350df1482e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-03-10 14:01:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c612eb15f06bfa7d9f0ded91fa49e195e030d0c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-10 14:09:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/37/536337/5/check/requirements-check/e0ad2a0/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/37/536337/5/check/openstack-tox-pep8/a2a16a3/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":5},{"id":"53a55c92c40348c357ffed383bfc161580c10e1e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-03-13 07:39:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"56a2358abce47d839f91b3c4c2eecb7d1c56c08c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-13 07:55:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/37/536337/6/check/requirements-check/067ab09/ : SUCCESS in 3m 04s\n- openstack-tox-pep8 http://logs.openstack.org/37/536337/6/check/openstack-tox-pep8/995af7c/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":6},{"id":"5860c43798722807bca5a6383761dd438722e182","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-03-14 06:14:30.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e181ecf79f39253fcae683a71873aa77b12e11cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-14 06:28:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/37/536337/7/check/requirements-check/834dc39/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/37/536337/7/check/openstack-tox-pep8/d0d6a20/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":7},{"id":"1d5bc69b9a4f32cedd5e2e9831b983211b0c1a95","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-03-15 00:39:45.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"6698e44d1cb89387d48f06dc2cbb3526c17350ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 00:56:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/37/536337/8/check/requirements-check/e647186/ : SUCCESS in 2m 55s\n- openstack-tox-pep8 http://logs.openstack.org/37/536337/8/check/openstack-tox-pep8/373a924/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":8},{"id":"83ead8cbe54e5a19723da8857043d49f7eb4d6fc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-03-15 09:41:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a1501c6bb2921dca63f518c6407158e18cddc179","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 09:53:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/37/536337/9/check/requirements-check/83854f5/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/37/536337/9/check/openstack-tox-pep8/468c5e2/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":9},{"id":"c79d806c4c99d61a2c139b852ea6c2ce94a9f6c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 16:13:05.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"e9ba45a3f49836f283ec90cd2163d6a7d7f037ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 17:10:58.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5290ef37aa274cf68756f8044e05f5ad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/37e57281e2ab4c6c831c8e074eca86b2 : SUCCESS in 1m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9d64b5f5417342a3831f1eb2fa91cb88 : SUCCESS in 1m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9715a18980a0465d9ba6aa3255868691 : SUCCESS in 2m 20s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/1b6a274e58304ba29341ce7568022779 : SUCCESS in 26m 49s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/55d65c88b1664dd6b895ad54116c55ee : SUCCESS in 21m 31s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/1c18a1eb95cd48efb44507d53b708e62 : SUCCESS in 26m 33s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/5f238b1719a74bfa90c8c03648fe97fd : SUCCESS in 25m 41s","accounts_in_message":[],"_revision_number":10},{"id":"5a1b9b991097d69f12f956c7e9a2ac04150a9093","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 19:19:48.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"6efcb6bd1906aab51b16e0ede2d5a261e3b71976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:20:19.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"82f0d64b2283113264c75fd23aada64f8352c551","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 20:01:09.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e67202ae7a94fa4b23d092f8fcbcaf1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/896b2fa615484c6eaedb068bbf1b993c : SUCCESS in 1m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ed55084fde06406cb1172a0fcce04313 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e951fecf22ec45acb5693240f92d58ca : SUCCESS in 2m 26s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/183af5f2e4fc498a9ff972c7ea865c55 : FAILURE in 26m 56s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/6a787acd7ee9486a91f3502725687a7b : SUCCESS in 35m 25s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/9c038bfe6d13479a98f770fd94a58ee8 : SUCCESS in 23m 22s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/16824f7faa464d039fdbc2ed9acdc973 : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":10},{"id":"97df852bd602da5234f5423b2c927c53fd0d79c6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-25 12:42:52.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f740ca351b24b73f823a5c8d02c166060935f522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 13:33:28.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e772bb4280f640dc86a842a1307a48ae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8c2ee30bff8c4c1e9847571d691c29cb : SUCCESS in 1m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/59f1be29b04d427ab7a3e7edce771a4a : SUCCESS in 1m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eba0ba4b1f546109c60aa1ebf5ca0cd : SUCCESS in 2m 37s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/a0e50c0b235145afb2001a4e727059e4 : SUCCESS in 32m 49s\n- zaqar-tempest-plugin-mongodb-2023-1 https://zuul.opendev.org/t/openstack/build/988abf517e0f4821bf19b5d61f230464 : FAILURE in 27m 25s\n- zaqar-tempest-plugin-mongodb-2023-2 https://zuul.opendev.org/t/openstack/build/207d9255df88463381350d1387ec322a : SUCCESS in 35m 42s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/00daf906d9ed461ebfbafe3c6bb192f8 : SUCCESS in 34m 13s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/d28867a85b514f89a247ca874ea8523c : SUCCESS in 39m 56s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/1b1aabdd13d7461687ec9ea72a583899 : SUCCESS in 22m 49s\n- zaqar-tempest-plugin-redis-2023-1 https://zuul.opendev.org/t/openstack/build/caeb5d45283944b5985a9fad6443705f : SUCCESS in 43m 57s\n- zaqar-tempest-plugin-redis-2023-2 https://zuul.opendev.org/t/openstack/build/46a2c34128b246bbaa5e35f89b8c61e3 : SUCCESS in 24m 31s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/9db934e1ec414befa9f57683aae2e18a : SUCCESS in 36m 51s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/b0a981b7ad9343c285ba9b5bcd585d7b : SUCCESS in 33m 57s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/9ec15efc09eb48e19d0f3d8dbe4936d1 : SUCCESS in 39m 22s\n- zaqar-tempest-plugin-swift-2023-1 https://zuul.opendev.org/t/openstack/build/c92a0a5c549140dda19027bd8924ee5b : SUCCESS in 26m 25s\n- zaqar-tempest-plugin-swift-2023-2 https://zuul.opendev.org/t/openstack/build/ab602d60f2144eb3be9458f2a5d35cb7 : SUCCESS in 34m 17s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/85adafc9f0d049929d906c95c042edd2 : SUCCESS in 27m 21s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/d3394e159b9d4f78a68bfd135fe79a64 : SUCCESS in 35m 01s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/c86c2d796a9341dc94444105e5877c78 : SUCCESS in 24m 32s","accounts_in_message":[],"_revision_number":10},{"id":"2e617530fe13d752d4bf1b9a67c8a079898a2094","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 03:04:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"435a11343f11bf172474762100c76260a9b9c3ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 03:53:52.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1c1ffc7d8244257ace706584d09bfbb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4cbe48e4a8b24dae96e695ff87617fab : SUCCESS in 2m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3db8bd649ea74229873e40c74c5f853d : SUCCESS in 1m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1540ab7d9bf4313a6fcee236fc07320 : SUCCESS in 3m 33s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/9635ff633f4847a7b217d9c2aff70630 : SUCCESS in 22m 12s\n- zaqar-tempest-plugin-mongodb-2023-1 https://zuul.opendev.org/t/openstack/build/24bf881290ca447e9a3ef6fd34dc18ab : SUCCESS in 25m 49s\n- zaqar-tempest-plugin-mongodb-2023-2 https://zuul.opendev.org/t/openstack/build/399e30333f5c4313b1e13dd2f88964af : SUCCESS in 24m 35s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/ee2a7ffbfa754d54a7c196d26f89761b : SUCCESS in 21m 25s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/562701d09ef0445484c6922dbdaa55ef : SUCCESS in 18m 24s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/4711eeda13aa49f1a031cdf2fe82b65e : SUCCESS in 34m 16s\n- zaqar-tempest-plugin-redis-2023-1 https://zuul.opendev.org/t/openstack/build/21260e529aca44d789acb53bdcefa8a4 : SUCCESS in 38m 45s\n- zaqar-tempest-plugin-redis-2023-2 https://zuul.opendev.org/t/openstack/build/b8d61c15f5394b608cd7c16af9f34829 : SUCCESS in 34m 47s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/4ce728c8ed0b4e09b6c589b9192341e5 : SUCCESS in 36m 58s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/9c3d0521b97c46d98249d5b3923ac837 : SUCCESS in 36m 05s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/34fd7410f83a4f39b30918edb4254470 : SUCCESS in 29m 29s\n- zaqar-tempest-plugin-swift-2023-1 https://zuul.opendev.org/t/openstack/build/145363f6e9fa4677a7cd5b9cbaaf2c89 : SUCCESS in 26m 43s\n- zaqar-tempest-plugin-swift-2023-2 https://zuul.opendev.org/t/openstack/build/2f7314c7797341eba1dc03be1ed596fc : SUCCESS in 37m 27s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/43dfb308ec8f4a579132a92322a14ede : SUCCESS in 38m 59s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/d166f3ece3a34ebdad9334c713e96dec : SUCCESS in 23m 49s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/a6faba2f132148189070fbccf2bccd12 : SUCCESS in 22m 20s","accounts_in_message":[],"_revision_number":10},{"id":"eb4be42941bffaa34956f9c071dd5e8a637cbc18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 03:54:26.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"388fc52e9abcbc4f9b3e93236ddadd880d05f252","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 04:40:29.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bbf781a3eec417d938df66923f83749\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f6d995c7834748d2924bd2c2f7007c13 : SUCCESS in 2m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/16ef66892baf43bebc49517877ec38fd : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6c256bbacbe4da6bc866bad8c509d03 : SUCCESS in 2m 22s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/9710869f8c14461e8429f4a250b2d2ba : FAILURE in 29m 07s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/866929b69d264b719ab9db86aee622a7 : SUCCESS in 20m 01s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/3d07cc82fbac4f18a00a9e625639dad1 : SUCCESS in 32m 09s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/d800f6f1091e4f4ba80f7655d74a4699 : SUCCESS in 36m 21s","accounts_in_message":[],"_revision_number":10},{"id":"00db580df13be1c661636fb30b09aca85f0628a1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 13:26:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6751699006f2ac0bc9d031a81f4ce381f573efc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 13:51:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0881fcadfb84b7c92d057616f5c208b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/577356db6dff40e687f0ba9391c4bfc8 : SUCCESS in 1m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/08e7d58e4c97424dbaac7e133f5a201f : SUCCESS in 1m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf0a277ad1d74819b71ae82a16741731 : SUCCESS in 1m 51s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/7dc0349980c44cdab53850673d6e2a42 : SUCCESS in 20m 22s\n- zaqar-tempest-plugin-mongodb-2023-1 https://zuul.opendev.org/t/openstack/build/18b64a0200ee41438e5dcf9da5501ed9 : SUCCESS in 19m 37s\n- zaqar-tempest-plugin-mongodb-2023-2 https://zuul.opendev.org/t/openstack/build/3a2d4648fe484f40b00bf260f3072239 : SUCCESS in 20m 19s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/6c60877d4f464f1d8a24432299b0b7f6 : SUCCESS in 22m 36s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/ddd3b730f0594037822d30cd3c61b099 : SUCCESS in 20m 33s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/ae4a29fc5dc24b0fadb3e535bb99d201 : SUCCESS in 20m 03s\n- zaqar-tempest-plugin-redis-2023-1 https://zuul.opendev.org/t/openstack/build/0b0f1b8f4bad4ac6a2bd3c90bc83207a : SUCCESS in 22m 37s\n- zaqar-tempest-plugin-redis-2023-2 https://zuul.opendev.org/t/openstack/build/c8cddd7a0da04c29b25853abff3b6d31 : SUCCESS in 22m 05s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/38a6448db9cd4f5483dfa2c3af386346 : SUCCESS in 21m 56s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/c17dfc9bbc2a4f2d9e2213ee85cc4939 : SUCCESS in 21m 11s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/68838629b5f0421cb7f71705486db7d9 : SUCCESS in 22m 05s\n- zaqar-tempest-plugin-swift-2023-1 https://zuul.opendev.org/t/openstack/build/4514c3bf17bc42318adc5b416c1f35ab : SUCCESS in 24m 27s\n- zaqar-tempest-plugin-swift-2023-2 https://zuul.opendev.org/t/openstack/build/c780c6c975734a869d5d4061b2d16ef7 : SUCCESS in 23m 20s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/c0a1f61249c14384b52e1c4168e916e7 : SUCCESS in 22m 27s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/8e2a3cd453154ac392488c9fe0fce4c8 : SUCCESS in 20m 20s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/45a2741d12c14d1da2698829b654372c : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":10},{"id":"bb35b341e9dfdd9a36cb3049bd97264deecf7d2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 13:52:19.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"3ada5a859f5d31c8b271a7787ebe31768f7bbe69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 14:31:31.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0cb72613a9a45a7a4b3cc96a200b2b4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3aced728e0084994b5e3679f17484781 : SUCCESS in 1m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/336befd05cb246779c0afda2733d5a39 : SUCCESS in 1m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b437094d811a480180dc8ec509543de1 : SUCCESS in 2m 20s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/644d851e357f4fc0b503d6a05e2df89f : SUCCESS in 27m 26s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/845b4ed19ed5445399bac53ed94bf313 : SUCCESS in 25m 00s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/a2972ed74f45414a9d4310bcc9bcb59f : SUCCESS in 34m 13s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/c01038fb7bcc4dba9c37f98edcf00859 : SUCCESS in 20m 45s","accounts_in_message":[],"_revision_number":10},{"id":"9a4d35e8abb7b9e0dce66bb754772d9ade76affe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 14:31:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"237bcc87f19fb072936e7ffb301d3068d89aa120","revisions":{"6eca66db0a5ea1bf224e7e91e6a1943bec004f7d":{"kind":"REWORK","_number":1,"created":"2018-01-22 11:21:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/536337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/37/536337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/1"}}},"commit":{"parents":[{"commit":"2c466738b2aeef9e7c14ddc9ebc1b006f7436447","subject":"Fix zaqar tempest plugin grenade ci failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/2c466738b2aeef9e7c14ddc9ebc1b006f7436447"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-01-22 11:21:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-01-22 11:21:52.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I26b839a61ea68202ac15e0c0ce00a141f1753e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/6eca66db0a5ea1bf224e7e91e6a1943bec004f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/6eca66db0a5ea1bf224e7e91e6a1943bec004f7d"}]},"branch":"refs/heads/master"},"a6bd752b77fba1137e6fcec6f6bf6bef4b32bc74":{"kind":"REWORK","_number":2,"created":"2018-01-24 01:42:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/536337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/37/536337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/2"}}},"commit":{"parents":[{"commit":"2c466738b2aeef9e7c14ddc9ebc1b006f7436447","subject":"Fix zaqar tempest plugin grenade ci failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/2c466738b2aeef9e7c14ddc9ebc1b006f7436447"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-01-24 01:42:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-01-24 01:42:54.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I26b839a61ea68202ac15e0c0ce00a141f1753e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/a6bd752b77fba1137e6fcec6f6bf6bef4b32bc74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/a6bd752b77fba1137e6fcec6f6bf6bef4b32bc74"}]},"branch":"refs/heads/master"},"f2cc33dbbbe4cbf2a7d03cda56885e8389a04e21":{"kind":"REWORK","_number":3,"created":"2018-01-24 02:31:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/536337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/37/536337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/3"}}},"commit":{"parents":[{"commit":"2c466738b2aeef9e7c14ddc9ebc1b006f7436447","subject":"Fix zaqar tempest plugin grenade ci failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/2c466738b2aeef9e7c14ddc9ebc1b006f7436447"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-01-24 02:31:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-01-24 02:31:46.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I26b839a61ea68202ac15e0c0ce00a141f1753e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/f2cc33dbbbe4cbf2a7d03cda56885e8389a04e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/f2cc33dbbbe4cbf2a7d03cda56885e8389a04e21"}]},"branch":"refs/heads/master"},"8a537667d05f94502b971757f8fbfe9e99851515":{"kind":"REWORK","_number":4,"created":"2018-01-27 18:51:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/536337/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/37/536337/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/4"}}},"commit":{"parents":[{"commit":"2c466738b2aeef9e7c14ddc9ebc1b006f7436447","subject":"Fix zaqar tempest plugin grenade ci failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/2c466738b2aeef9e7c14ddc9ebc1b006f7436447"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-01-27 18:51:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-01-27 18:51:44.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I26b839a61ea68202ac15e0c0ce00a141f1753e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/8a537667d05f94502b971757f8fbfe9e99851515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/8a537667d05f94502b971757f8fbfe9e99851515"}]},"branch":"refs/heads/master"},"26348c49fad18d0770382d237bacc4119f07cc76":{"kind":"REWORK","_number":5,"created":"2018-03-10 14:01:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/536337/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/37/536337/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/5"}}},"commit":{"parents":[{"commit":"92d57f52a1890a51d89b3c10918e77fcd440ac05","subject":"Tempest plugin support for message body md5.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/92d57f52a1890a51d89b3c10918e77fcd440ac05"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-10 14:01:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-10 14:01:13.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I26b839a61ea68202ac15e0c0ce00a141f1753e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/26348c49fad18d0770382d237bacc4119f07cc76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/26348c49fad18d0770382d237bacc4119f07cc76"}]},"branch":"refs/heads/master"},"1dd9f4bd55f6cc9e13e27f8026efc304a4f5fd9d":{"kind":"REWORK","_number":6,"created":"2018-03-13 07:39:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/536337/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/37/536337/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/6"}}},"commit":{"parents":[{"commit":"92d57f52a1890a51d89b3c10918e77fcd440ac05","subject":"Tempest plugin support for message body md5.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/92d57f52a1890a51d89b3c10918e77fcd440ac05"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-13 07:39:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-13 07:39:11.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I26b839a61ea68202ac15e0c0ce00a141f1753e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/1dd9f4bd55f6cc9e13e27f8026efc304a4f5fd9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/1dd9f4bd55f6cc9e13e27f8026efc304a4f5fd9d"}]},"branch":"refs/heads/master"},"a626e64fc4823f20a928df8c79d54beea83736fc":{"kind":"REWORK","_number":7,"created":"2018-03-14 06:14:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/536337/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/37/536337/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/7"}}},"commit":{"parents":[{"commit":"92d57f52a1890a51d89b3c10918e77fcd440ac05","subject":"Tempest plugin support for message body md5.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/92d57f52a1890a51d89b3c10918e77fcd440ac05"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-14 06:14:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-14 06:14:28.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I26b839a61ea68202ac15e0c0ce00a141f1753e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/a626e64fc4823f20a928df8c79d54beea83736fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/a626e64fc4823f20a928df8c79d54beea83736fc"}]},"branch":"refs/heads/master"},"48dd7c513891ff2bf907d64bb4e429a5b3ddcdf8":{"kind":"NO_CHANGE","_number":8,"created":"2018-03-15 00:39:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/536337/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/37/536337/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/8"}}},"commit":{"parents":[{"commit":"92d57f52a1890a51d89b3c10918e77fcd440ac05","subject":"Tempest plugin support for message body md5.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/92d57f52a1890a51d89b3c10918e77fcd440ac05"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-15 00:39:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-15 00:39:42.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I26b839a61ea68202ac15e0c0ce00a141f1753e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/48dd7c513891ff2bf907d64bb4e429a5b3ddcdf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/48dd7c513891ff2bf907d64bb4e429a5b3ddcdf8"}]},"branch":"refs/heads/master"},"37534ba4052d7639d665b71c4c575dc2ffdeea16":{"kind":"REWORK","_number":9,"created":"2018-03-15 09:41:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/536337/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/37/536337/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/9"}}},"commit":{"parents":[{"commit":"9051907cf70527b70763ae39b096b6ba477c3096","subject":"fix error url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/9051907cf70527b70763ae39b096b6ba477c3096"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-15 09:40:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-15 09:40:59.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I26b839a61ea68202ac15e0c0ce00a141f1753e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/37534ba4052d7639d665b71c4c575dc2ffdeea16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/37534ba4052d7639d665b71c4c575dc2ffdeea16"}]},"branch":"refs/heads/master"},"237bcc87f19fb072936e7ffb301d3068d89aa120":{"kind":"REWORK","_number":10,"created":"2024-10-24 16:13:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/536337/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/37/536337/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/37/536337/10"}}},"commit":{"parents":[{"commit":"96a4ea24b7f6e8ad3ebb39509648ed3397d5f15a","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/96a4ea24b7f6e8ad3ebb39509648ed3397d5f15a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-15 09:40:59.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 16:12:34.000000000","tz":540},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I26b839a61ea68202ac15e0c0ce00a141f1753e02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/237bcc87f19fb072936e7ffb301d3068d89aa120"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/237bcc87f19fb072936e7ffb301d3068d89aa120"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
