)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~707323","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~I0fca9cb65b65da6d8f386175fc5c29abf12c4311","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"bp/query-queues-with-count","hashtags":[],"change_id":"I0fca9cb65b65da6d8f386175fc5c29abf12c4311","subject":"Add tempest test for querying queus with count","status":"MERGED","created":"2020-02-12 06:59:06.000000000","updated":"2020-02-23 19:13:01.000000000","submitted":"2020-02-14 07:43:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707323-1581666233806-b4850ae3","meta_rev_id":"592df5590f86896fc924c6cffac5ba244a7004e0","_number":707323,"virtual_id_number":707323,"owner":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 07:43:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2020-02-13 01:20:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-13 03:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-13 03:35:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 01:20:43.000000000","updated_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2020-02-14 07:43:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fde9cfcfc70ded37908fd12395438f58417ee0f","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2020-02-12 06:59:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da72cf0971a127d6af002000d0dad0e02d5bb71c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 07:08:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/292f7d9d4480400cad0d88d5c7bf6af3 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b5a24400a79f044c7b30c4785ba7124d3c887c7c","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2020-02-12 07:39:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"833924faedfe1db4c4d801e935877c20e814c388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 07:45:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7804c1e9224015855b1cac9996c99d : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"2dbaf21231ff65146e1a752a3a7d01e8b24b5f6d","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2020-02-12 14:10:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"588e58fdc9deb34b7c7657b900b46f686ecce38a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 14:16:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3085818d9c1490781a8a51a9aa2d497 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"e1f77b729a3b7251bbf1968b6fd9e2e5d2c3dc5d","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2020-02-13 01:20:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a11b007fc2cf79dc4ffaed9a4948803a60c7e9cd","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2020-02-13 03:35:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ce56f665e860e5c19a85e5f2abc3d630cc67dcbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 03:35:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"207ea606c9fa7eee11e6166163d59a07bf3c174c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 03:40:53.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfce42f0abd94ec6a94f51810b17ee04 : RETRY_LIMIT in 1s","accounts_in_message":[],"_revision_number":3},{"id":"820ecf547e48293f1bec57e0edc536c25fcd78ef","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2020-02-14 07:23:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cf3ff0c9a73b39eac107ff8fafa85857079fd255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 07:34:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40a0b60b98c54d7992aa6e84c12fe3ef : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":3},{"id":"bba725a3691c7885e6a88de8f192e7eb7a07c039","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 07:34:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"55455221095db1b2977df3f9179c4e117402f29e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 07:43:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3a7b4ff87fbc97caa0964520cf012507674a3a37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 07:43:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4e2c416ffee4661b5ca46c0d22eb5e7 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"02246f0a4389e5f6dd95d892467fd3551a4069b0","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-02-14 16:50:50.000000000","message":"Patch Set 3:\n\nthis new test is failing in our jobs for centos7 and centos8 jobs, is this tested somewhere upstream:-\nhttps://logserver.rdoproject.org/ci.centos.org/weirdo-generic-puppet-openstack-scenario002/13627/weirdo-project/logs/testr_results.html\nhttps://logserver.rdoproject.org/ci.centos.org/weirdo-generic-puppet-openstack-scenario002/13621/weirdo-project/logs/testr_results.html","accounts_in_message":[],"_revision_number":3},{"id":"8b2ca1b3c42b47544c083bdb669439670e6d67ee","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-02-17 06:55:57.000000000","message":"Patch Set 3:\n\n\u003e this new test is failing in our jobs for centos7 and centos8 jobs,\n \u003e is this tested somewhere upstream:-\n \u003e https://logserver.rdoproject.org/ci.centos.org/weirdo-generic-puppet-openstack-scenario002/13627/weirdo-project/logs/testr_results.html\n \u003e https://logserver.rdoproject.org/ci.centos.org/weirdo-generic-puppet-openstack-scenario002/13621/weirdo-project/logs/testr_results.html\n\nso zaqar gates are also broken with https://review.opendev.org/#/c/708060/","accounts_in_message":[],"_revision_number":3},{"id":"2186d18aa6fbd5633f27f7e5bd238403cec672e3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 19:13:01.000000000","message":"Created a revert of this change as Ib946f57855ac22948792eaaa2bb901ad46704779","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"04f720906d5b75b8efdcda29203086990efa6c75","revisions":{"9f2ede84a0dafca9b3fdccbde8cd40294723ab9e":{"kind":"REWORK","_number":1,"created":"2020-02-12 06:59:06.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/23/707323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/23/707323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/1"}}},"commit":{"parents":[{"commit":"ad91e171db3f052af3087c01aca0e6d91bff3265","subject":"Fix incompatibility in python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/ad91e171db3f052af3087c01aca0e6d91bff3265"}]}],"author":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2020-02-12 06:58:25.000000000","tz":480},"committer":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2020-02-12 06:58:25.000000000","tz":480},"subject":"Add tempest test for querying queus with count","message":"Add tempest test for querying queus with count\n\nChange-Id: I0fca9cb65b65da6d8f386175fc5c29abf12c4311\nImplements: blueprint query-queues-with-count\nSigned-off-by: wanghao \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/9f2ede84a0dafca9b3fdccbde8cd40294723ab9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/9f2ede84a0dafca9b3fdccbde8cd40294723ab9e"}]},"branch":"refs/heads/master"},"abf80876749ca2a1aa11dbdcb2881113f3816cc5":{"kind":"REWORK","_number":2,"created":"2020-02-12 07:39:04.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/23/707323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/23/707323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/2"}}},"commit":{"parents":[{"commit":"ad91e171db3f052af3087c01aca0e6d91bff3265","subject":"Fix incompatibility in python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/ad91e171db3f052af3087c01aca0e6d91bff3265"}]}],"author":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2020-02-12 06:58:25.000000000","tz":480},"committer":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2020-02-12 07:38:54.000000000","tz":480},"subject":"Add tempest test for querying queus with count","message":"Add tempest test for querying queus with count\n\nChange-Id: I0fca9cb65b65da6d8f386175fc5c29abf12c4311\nImplements: blueprint query-queues-with-count\nSigned-off-by: wanghao \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/abf80876749ca2a1aa11dbdcb2881113f3816cc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/abf80876749ca2a1aa11dbdcb2881113f3816cc5"}]},"branch":"refs/heads/master"},"04f720906d5b75b8efdcda29203086990efa6c75":{"kind":"REWORK","_number":3,"created":"2020-02-12 14:10:49.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/23/707323/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/23/707323/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/707323/3"}}},"commit":{"parents":[{"commit":"ad91e171db3f052af3087c01aca0e6d91bff3265","subject":"Fix incompatibility in python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/ad91e171db3f052af3087c01aca0e6d91bff3265"}]}],"author":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2020-02-12 06:58:25.000000000","tz":480},"committer":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2020-02-12 14:10:26.000000000","tz":480},"subject":"Add tempest test for querying queus with count","message":"Add tempest test for querying queus with count\n\nChange-Id: I0fca9cb65b65da6d8f386175fc5c29abf12c4311\nImplements: blueprint query-queues-with-count\nSigned-off-by: wanghao \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/04f720906d5b75b8efdcda29203086990efa6c75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/04f720906d5b75b8efdcda29203086990efa6c75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
