)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~731424","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~Ifdb0f91090de529de095c48d87fe017b92a0b78f","project":"openstack/zaqar-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ifdb0f91090de529de095c48d87fe017b92a0b78f","subject":"Add py38 package metadata","status":"ABANDONED","created":"2020-05-28 03:59:27.000000000","updated":"2020-06-12 03:09:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2ee13ec8e950a978689223fb7c4d91bdd534d156","_number":731424,"virtual_id_number":731424,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-28 04:08:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 04:08:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"751fadb3a3db7b1fb09c81692e3c64f219f6ef45","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2020-05-28 03:59:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53f095e6bd22addc22149c7823bf802db7c5d4ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 04:08:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/823020d9d48a47228002358b4b100cfe : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5882303032204c7aae9bca8bbc36811b : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2ee13ec8e950a978689223fb7c4d91bdd534d156","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2020-06-12 03:09:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d747f7e52c4f06d3aa002afd7744fda29aa4e8ea","revisions":{"d747f7e52c4f06d3aa002afd7744fda29aa4e8ea":{"kind":"REWORK","_number":1,"created":"2020-05-28 03:59:27.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/24/731424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/24/731424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/24/731424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/24/731424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/24/731424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/24/731424/1"}}},"commit":{"parents":[{"commit":"ef3aee125090ebf0ae56b65cdc86535d355cb9d4","subject":"Optimize the tempest job for zaqar-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/ef3aee125090ebf0ae56b65cdc86535d355cb9d4"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2020-05-28 03:24:53.000000000","tz":480},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2020-05-28 03:59:18.000000000","tz":480},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nChange-Id: Ifdb0f91090de529de095c48d87fe017b92a0b78f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/d747f7e52c4f06d3aa002afd7744fda29aa4e8ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/d747f7e52c4f06d3aa002afd7744fda29aa4e8ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
