)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~744638","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~I5da8ed8555fb5b01c88f7f7ca9c6c267e2e0eb29","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I5da8ed8555fb5b01c88f7f7ca9c6c267e2e0eb29","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-04 03:34:18.000000000","updated":"2020-08-04 15:44:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0acf64f6f0e08ab2c61e3b6c65fba18a1d1392cc","_number":744638,"virtual_id_number":744638,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-04 05:20:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 05:20:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a9175e57537aaa98174389451a59531aa8cd8fa","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 03:34:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba02375113b82c0c6d8af7cfd64db34dec93e85e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 05:20:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f71ff3ed1cb34365a62f26e7d487b569 : SUCCESS in 3m 58s\n- zaqar-tempest-plugin-tempest-redis https://zuul.opendev.org/t/openstack/build/eadabad8ba894653aa3d779afb537bb4 : SUCCESS in 18m 17s\n- zaqar-tempest-plugin-tempest-swift https://zuul.opendev.org/t/openstack/build/e74b0a8ae6a8400a90813bda31aa0590 : SUCCESS in 20m 54s\n- zaqar-tempest-plugin-tempest-mongodb https://zuul.opendev.org/t/openstack/build/74dac8001a54400aa282f9e00771d53f : SUCCESS in 18m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d7662ded3a102c76f60d020bd88135285d93bb1f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 15:44:55.000000000","message":"Abandoned\n\nall working fine, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b23af5f1ce782a19b831ef42cc98272f1da0f63","revisions":{"2b23af5f1ce782a19b831ef42cc98272f1da0f63":{"kind":"REWORK","_number":1,"created":"2020-08-04 03:34:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/38/744638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/38/744638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/744638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/744638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/744638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/744638/1"}}},"commit":{"parents":[{"commit":"4195ae11b22404f757e0e21f2a0452524152163f","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/4195ae11b22404f757e0e21f2a0452524152163f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 03:21:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 03:34:17.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40229\n\nDepends-On: https://review.opendev.org/#/c/734700/\nChange-Id: I5da8ed8555fb5b01c88f7f7ca9c6c267e2e0eb29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/2b23af5f1ce782a19b831ef42cc98272f1da0f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/2b23af5f1ce782a19b831ef42cc98272f1da0f63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
