)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~769789","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~I2aebaf6be3ffc86659782f8272f03a2ea0463c8c","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"fix-relmgt-pip-doc","hashtags":[],"change_id":"I2aebaf6be3ffc86659782f8272f03a2ea0463c8c","subject":"Add doc/requirements","status":"MERGED","created":"2021-01-07 16:29:16.000000000","updated":"2021-01-15 08:14:15.000000000","submitted":"2021-01-15 08:14:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769789-fix-relmgt-pip-doc","meta_rev_id":"02767c92d967a58208325e7404367916c54e5b9a","_number":769789,"virtual_id_number":769789,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 08:14:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-15 07:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-15 07:18:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-07 17:25:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-15 07:18:27.000000000","updated_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"reviewer":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"state":"REVIEWER"}],"messages":[{"id":"cd23f8ed20f6387fa98b1741afbb7359e252c39a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-07 16:29:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b8d175ed26884b69415c3a5c7c4f075ffc84368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 17:25:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9cebeece3de44ef3bb77896a3f532253 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aed5b078c63c47e284a1c4aeb99c015a : SUCCESS in 3m 05s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/6786d0323d894848a932724cef4af4c7 : SUCCESS in 42m 12s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/460365ac9499401288440e3652786963 : SUCCESS in 46m 12s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/8073e413020a4f0ca121081397807145 : SUCCESS in 42m 13s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/68ba842e320140dc9014320a6d1e53a6 : SUCCESS in 49m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5bb37f82059bb7f6bd9f52ee814149965c3de751","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2021-01-15 07:18:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b72634a30f4882970253df206780caeb82dc0063","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 07:18:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41518240baa0005a8578f77fa73893abecd3870d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:14:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bc87a1585b8245638dbd740b918d9b40 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d1551bc81541cc807d307ef719e18a : SUCCESS in 4m 18s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/f1bd8c3bafbc4e7fb8d4cf701053c4c8 : SUCCESS in 52m 52s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/b9a2b068f6964a75946c51d0f4becb23 : SUCCESS in 52m 44s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/ad52bbf5168e417a896ef18094ff0a91 : SUCCESS in 36m 48s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/0f1afc65e6504461bf1a6c377c9664b8 : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":1},{"id":"02767c92d967a58208325e7404367916c54e5b9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:14:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fef59164aafef3f37191e78692713219f8de4bff","revisions":{"fef59164aafef3f37191e78692713219f8de4bff":{"kind":"REWORK","_number":1,"created":"2021-01-07 16:29:16.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/769789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/89/769789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/89/769789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/89/769789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/89/769789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/89/769789/1"}}},"commit":{"parents":[{"commit":"9146cc9dbfcb7c6c207cf1660288ec03e172ff4e","subject":"New native Zuul v3 version of the tempest jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/9146cc9dbfcb7c6c207cf1660288ec03e172ff4e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-07 16:29:03.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-07 16:29:03.000000000","tz":60},"subject":"Add doc/requirements","message":"Add doc/requirements\n\nWe need to specify doc requirements in doc/requirements.txt\nto avoid problems with the pip resolver [1] for the release team [2][3].\nRemoving specific doc requirements from test-requirements.txt.\n\nThe problem here is that this repos haven\u0027t doc/requirements.txt file\nand by default in this case zuul will use the test-requirements.txt file\nto pull requirements [4].\n\nThis requirements file contains extra requirements like flake8 that\ncollided with those allowed in our job environment and so the new pip\nresolver fails to install these requirements and the job exits in error.\n\nThis project meet the conditions leading to the bug.\n\n[1] http://lists.openstack.org/pipermail/release-job-failures/2021-January/001500.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019611.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019612.html\n[4] https://opendev.org/zuul/zuul-jobs/src/branch/master/roles/ensure-sphinx/tasks/main.yaml#L36\n\nChange-Id: I2aebaf6be3ffc86659782f8272f03a2ea0463c8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/fef59164aafef3f37191e78692713219f8de4bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/fef59164aafef3f37191e78692713219f8de4bff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
