)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~801000","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~Ifd6fc7b0c22919b91fff9eb136de8c023bce3849","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"stestr","hashtags":[],"change_id":"Ifd6fc7b0c22919b91fff9eb136de8c023bce3849","subject":"Migrate from testr to stestr","status":"MERGED","created":"2021-07-15 21:58:12.000000000","updated":"2021-07-26 01:53:05.000000000","submitted":"2021-07-26 01:53:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"801000-stestr","meta_rev_id":"07a11c4018c6ca120b4094cf304eeb025826a97e","_number":801000,"virtual_id_number":801000,"owner":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-26 01:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"recommended":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2021-07-22 17:44:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-26 01:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-26 01:16:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"reviewers":{"REVIEWER":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-15 22:04:02.000000000","updated_by":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-07-15 22:43:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-23 20:58:33.000000000","updated_by":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"reviewer":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"state":"REVIEWER"},{"updated":"2021-07-23 21:02:36.000000000","updated_by":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2021-07-24 09:28:48.000000000","updated_by":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"}],"messages":[{"id":"32dc4240b681ebd5e0262ac84ad9b5d18810e981","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"date":"2021-07-15 21:58:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ce29e0e5100f1f668d94e956e7c6d0504035f54","author":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"date":"2021-07-15 22:04:02.000000000","message":"Patch Set 1:\n\nHey Martin, \nPlease review my code.","accounts_in_message":[],"_revision_number":1},{"id":"cec3ab5a3efaded072f918ec9d288622319a6461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 22:43:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/40b4690a98654ab58ba34c4a92a24c96 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/104452db87784af78c5238f0ab20cb06 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e1219ee664b42428c67b554d62a474e : FAILURE in 3m 53s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/355f1924dbfe4de59aff70c9092c6d6f : SUCCESS in 35m 49s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/32ef6a963542444ba4f045a93eca8b85 : SUCCESS in 40m 25s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/31324e380c5444a993c7c5fc155dde67 : SUCCESS in 37m 46s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/876e597369c341d9952a687eaac2df2f : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b2e2e5ab9b04396c448ae0b2b0bfe5bb0e2196cb","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-07-17 17:39:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"872bbfda33d82daaa23d76b012920057847ee023","author":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"date":"2021-07-20 12:24:52.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nHey Martin, Should I remove \"openstack-cover-jobs\" line, and then send for review?","accounts_in_message":[],"_revision_number":1},{"id":"cfef3ff26058cf8239c823fc204743b876a871f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"date":"2021-07-22 08:14:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f1c1dfa6d0ed1ee849a29c49f7e8bcfbc41467e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 08:54:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/95580f89547141e98fda1a59145d4774 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcf63d2bbc48408290a8e5702d5998f9 : SUCCESS in 4m 36s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/49c32691d8d14a05895d0fed975a21b3 : SUCCESS in 35m 40s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/9a02cb5ff15b4ed59a48e3fd84d5d6db : SUCCESS in 37m 55s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/86c320b2e0614a0ba4556d2968a0f65e : SUCCESS in 35m 32s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/ea400946da9f455abf3bf1d45a8d9263 : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":2},{"id":"cbb7a65abbcfee5a436767ac73f33e11f9212896","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-07-22 17:44:00.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"e514eda7368afdad6ac7c93005af5dc18306f420","author":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"date":"2021-07-22 18:02:06.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Patch Set 2: Code-Review+1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e thanks","accounts_in_message":[],"_revision_number":2},{"id":"a4107d8c3628050c3189fb9a0a225d61ac877143","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-07-22 19:36:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"418b22919717bdce72777c00fa4fa0c3e3078016","author":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"date":"2021-07-23 20:58:33.000000000","message":"Patch Set 2:\n\nHey Hao Wang,\nPlease review my code.","accounts_in_message":[],"_revision_number":2},{"id":"331c14b1a2fa63ef64471e5eb190e5917b5353d5","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2021-07-26 01:15:34.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM, thanks.","accounts_in_message":[],"_revision_number":2},{"id":"42ea7b42f54dbda20b6af4ade4a57022b556cf0e","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2021-07-26 01:16:19.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6601d564e168e21103807d4631807206f86eca2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 01:16:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"069948debf6b8cd338268a70ba1d0c4dfd1e2899","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 01:53:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/830e9abb1f0249869be14c878d4def34 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74cf75a4da6b4daa96be716be09da992 : SUCCESS in 3m 31s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/3ace9fbc76d049daaecfba1d71a833bb : SUCCESS in 33m 53s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/c607255b00c44965b9679837728ea6ef : SUCCESS in 30m 31s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/0a3c5ab16be94e88b81b537d3f299b22 : SUCCESS in 35m 38s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/4c3c50b776e945e59e152486222303ea : SUCCESS in 34m 15s","accounts_in_message":[],"_revision_number":2},{"id":"07a11c4018c6ca120b4094cf304eeb025826a97e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 01:53:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3813c99e501a1f67a9cbde751cf71747a61f6786","revisions":{"ea675908489287d534d5214576e5f9da1e458984":{"kind":"REWORK","_number":1,"created":"2021-07-15 21:58:12.000000000","uploader":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"ref":"refs/changes/00/801000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/00/801000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/00/801000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/00/801000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/00/801000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/00/801000/1"}}},"commit":{"parents":[{"commit":"fef59164aafef3f37191e78692713219f8de4bff","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/fef59164aafef3f37191e78692713219f8de4bff"}]}],"author":{"name":"JyotiJauhari","email":"jyotijauhari222@gmail.com","date":"2021-07-15 21:48:27.000000000","tz":330},"committer":{"name":"JyotiJauhari","email":"jyotijauhari222@gmail.com","date":"2021-07-15 21:53:08.000000000","tz":330},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\n* Replace .testr.conf by .stestr.conf for migration and update\n  .gitignore and test-requirements.txt file accordingly\n* Use py3 as the default runtime for tox\n* Add a new job, openstack-cover-jobs, to run the coverage in Zuul\n* Use tox version 3.18.0 for using allowlist_externals\n\nSigned-off-by: JyotiJauhari \u003cjyotijauhari222@gmail.com\u003e\nChange-Id: Ifd6fc7b0c22919b91fff9eb136de8c023bce3849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/ea675908489287d534d5214576e5f9da1e458984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/ea675908489287d534d5214576e5f9da1e458984"}]},"branch":"refs/heads/master"},"3813c99e501a1f67a9cbde751cf71747a61f6786":{"kind":"REWORK","_number":2,"created":"2021-07-22 08:14:11.000000000","uploader":{"_account_id":33766,"name":"Jyoti Jauhari","email":"jyotijauhari222@gmail.com","username":"techyexito"},"ref":"refs/changes/00/801000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/00/801000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/00/801000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/00/801000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/00/801000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/00/801000/2"}}},"commit":{"parents":[{"commit":"fef59164aafef3f37191e78692713219f8de4bff","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/fef59164aafef3f37191e78692713219f8de4bff"}]}],"author":{"name":"JyotiJauhari","email":"jyotijauhari222@gmail.com","date":"2021-07-15 21:48:27.000000000","tz":330},"committer":{"name":"JyotiJauhari","email":"jyotijauhari222@gmail.com","date":"2021-07-22 08:02:01.000000000","tz":330},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\n* Replace .testr.conf by .stestr.conf for migration and update\n  .gitignore and test-requirements.txt file accordingly\n* Use py3 as the default runtime for tox\n* Add a new job, openstack-cover-jobs, to run the coverage in Zuul\n* Use tox version 3.18.0 for using allowlist_externals\n\nSigned-off-by: JyotiJauhari \u003cjyotijauhari222@gmail.com\u003e\nChange-Id: Ifd6fc7b0c22919b91fff9eb136de8c023bce3849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/3813c99e501a1f67a9cbde751cf71747a61f6786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/3813c99e501a1f67a9cbde751cf71747a61f6786"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
