)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~909323","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~Ide365a91b434c015cf44c9aa900813237b752be9","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-21 01:18:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8846":{"account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"last_update":"2024-02-20 19:13:21.000000000","reason":"\u003cGERRIT_ACCOUNT_8846\u003e replied on the change","reason_account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}}},"hashtags":[],"change_id":"Ide365a91b434c015cf44c9aa900813237b752be9","subject":"Remove six","status":"MERGED","created":"2024-02-18 16:03:53.000000000","updated":"2024-02-21 02:30:32.000000000","submitted":"2024-02-21 02:30:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909323-remove-six","meta_rev_id":"643a7f7e6e6c13125649091b483190a046ccd33a","_number":909323,"virtual_id_number":909323,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-21 02:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-20 19:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-20 19:13:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-19 13:40:18.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"state":"REVIEWER"},{"updated":"2024-02-20 01:54:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0539a9a9c94189d6163f4af35f4eb463c1b51190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-18 16:03:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d987cd46729f842743512e6e3dbbd1883458fcbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 01:05:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"9d1fbe9bc5cca802765f01bb19871bb420484daf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 01:54:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb55ba363d66417a9b05e71f677061cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1109d884c09640ec918401a9d80dabab : SUCCESS in 2m 27s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/853e5e36cca1404ebf7f62bd534441c7 : SUCCESS in 38m 05s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/f948e851fb254adf86464f8799903ae7 : SUCCESS in 31m 53s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/64307e90a2824c21bd53f7a2d3195f0b : SUCCESS in 26m 18s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/a3b1f1e62efb4c5a874245e8a7e62ae1 : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":2},{"id":"1276987383c4122fc94e5630f51805bc33cb5dda","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2024-02-20 19:13:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dcfcb4fadaa27d8617adf1008495c78ff0b7aa4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 19:17:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7ebb9459d354599ac0915e4af3e0066192acb183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 20:08:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb34037a14684fa48fb37960b97f0021\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07fdbfc05a7e4af2a4df098b5e7cf968 : SUCCESS in 2m 37s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/a15b61be4f704604b5a4460dd65f43b5 : FAILURE in 30m 30s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/adf66319add1461f8593012fbb14bfd3 : SUCCESS in 37m 30s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/7bfc6fa0bbe0420eaee678490c43ebb9 : SUCCESS in 28m 47s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/7ab87b5b8032482ebb82093344a9c016 : SUCCESS in 30m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2e48545fda859f6b182deb8e7cd4c4db72e38dc8","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2024-02-20 22:52:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d397512c3317daa94ef38024c22044a1fd41520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 23:38:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/905c7a7ea4284642bce041040ad3c1b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54b210d78a3744b691e7e01ce3d1796f : SUCCESS in 3m 42s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/702befeec10846c7967a61bcf6c8c923 : SUCCESS in 29m 03s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/9a7749c7b0c84f6187451e855fb2277d : SUCCESS in 38m 20s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/a167db914eb2447db3cb29260486d8ec : SUCCESS in 33m 29s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/4d7e934c9a3b42e99f39e90f47927fbd : SUCCESS in 40m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a389d62b7af449f5e7db0d3a61363924afd03aa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 23:38:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"43bc718fe2b7025d5058f22fb30145a8e5be57f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 00:25:44.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35d9fe60ec8945be994bfb71b98a6671\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea4b92edbbfc45eba21bcf6c5a732ece : SUCCESS in 2m 37s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/f475567b708f4ffea33e3b11a01d100e : FAILURE in 30m 07s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/5c3031077bf74743959b198036437948 : SUCCESS in 38m 43s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/3516928a78834ed9b165f97115cf977a : SUCCESS in 24m 19s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/638e8330acdf4961a9ecefcc350cb4e9 : SUCCESS in 24m 41s","accounts_in_message":[],"_revision_number":2},{"id":"52f4b382a6b696e19e75a19ff1e25a5cb4c8664e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 01:18:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d290490175e9103aa3e5729aef4f80a0393d492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 01:45:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17ec63fefa3a4ec59266a4f23ca90834\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0183bfde2e84b45a840c55733ea3112 : SUCCESS in 2m 10s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/f1c60101ca204b4ca3c330e3f6b157ff : SUCCESS in 23m 17s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/3501d9706f784b6ea2cfc01f17035872 : SUCCESS in 21m 58s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/eb1601dfa3244278a5508603f4da2b99 : SUCCESS in 25m 31s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/9bcf028d5ead41c8a0f3ecdd4d791dcc : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":2},{"id":"99babd37ab83d3f6d66ecdc1358eba59770dca47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 01:45:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c730a758bf5004d4a50fb3caea9fa87b52de90d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 02:30:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7117d99e96204dc7a2bdabf00c740a92\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ab061cc26e64aa9b7a1764bebd02317 : SUCCESS in 3m 25s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/45a5bbbd2e7a45578d843eb83b826435 : SUCCESS in 24m 31s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/ec84051f33384c8899efa60724d978b7 : SUCCESS in 36m 43s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/ac1b6d57fcf44c04af60ed8f0ae53a7a : SUCCESS in 27m 32s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/9dc21e57b1344fbfaf2ac60acec87855 : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":2},{"id":"643a7f7e6e6c13125649091b483190a046ccd33a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 02:30:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e0ed95af66ac09c292d59c9b6819ec63830da4ed","revisions":{"87fea65552a7022026f5622701f785f86cbe5e3b":{"kind":"REWORK","_number":1,"created":"2024-02-18 16:03:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/909323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/23/909323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/909323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/909323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/909323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/909323/1"}}},"commit":{"parents":[{"commit":"3813c99e501a1f67a9cbde751cf71747a61f6786","subject":"Migrate from testr to stestr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/3813c99e501a1f67a9cbde751cf71747a61f6786"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-18 16:03:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-18 16:03:50.000000000","tz":540},"subject":"Remove six","message":"Remove six\n\nThis plugin no longer supports python 2 thus usage of six is no longer\nneeded.\n\nChange-Id: Ide365a91b434c015cf44c9aa900813237b752be9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/87fea65552a7022026f5622701f785f86cbe5e3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/87fea65552a7022026f5622701f785f86cbe5e3b"}]},"branch":"refs/heads/master"},"e0ed95af66ac09c292d59c9b6819ec63830da4ed":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-20 01:05:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/909323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/23/909323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/909323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/909323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/909323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/23/909323/2"}}},"commit":{"parents":[{"commit":"026f0e732ec9739607c28c27c0774ff9b02895d6","subject":"zuul: Declare queue at top level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/026f0e732ec9739607c28c27c0774ff9b02895d6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-18 16:03:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 01:05:07.000000000","tz":540},"subject":"Remove six","message":"Remove six\n\nThis plugin no longer supports python 2 thus usage of six is no longer\nneeded.\n\nChange-Id: Ide365a91b434c015cf44c9aa900813237b752be9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/e0ed95af66ac09c292d59c9b6819ec63830da4ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/e0ed95af66ac09c292d59c9b6819ec63830da4ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
