)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~932899","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~I58cf063dc21b1a23313c00862c28551dec4cc8eb","project":"openstack/zaqar-tempest-plugin","branch":"master","hashtags":[],"change_id":"I58cf063dc21b1a23313c00862c28551dec4cc8eb","subject":"Skip tempest jobs for non-functional change","status":"MERGED","created":"2024-10-21 12:13:18.000000000","updated":"2024-10-24 14:50:28.000000000","submitted":"2024-10-24 14:50:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932899","meta_rev_id":"0f72d535e00c1fe8362baa523dbf227f2803b509","_number":932899,"virtual_id_number":932899,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-24 14:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-10-24 13:59:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-10-24 13:59:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-21 12:52:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f5282cd811c592d458c93fa3ffc559badb22792","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-21 12:13:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9268c5b7c87b8879fb06fc8fc7ab6a044566cfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 12:52:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3117ac5b6df748ac978abdd4a7408123\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2019d5dd187b4097b2ec824171f7a9b1 : SUCCESS in 2m 31s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/323e4aab0c754105843baf17b8993495 : SUCCESS in 32m 48s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/444def97a0e7407396a88309d60cab1f : SUCCESS in 31m 26s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/1907b9853fd7469a9eaa4f5a5fb8f8ae : SUCCESS in 35m 40s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/8b5ef500ecf141c3a7aa46e737937f3e : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3a9e1a7f9866aa8cc177eee133999325478522c6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 13:59:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"74345358db0c376707dab81b1743406e5e9d7175","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 14:00:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60c7d4003937eb57af17bb1e71269f858e3a285a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 14:50:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26fbab4e018d4caba52219fde31f8e75\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eab848b802c4ea5b0c309e6be4d5c21 : SUCCESS in 3m 50s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/c94ac3b9f7634426a5f86c91e359aca0 : SUCCESS in 21m 27s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/716c484884704442bb9393723e84fe91 : SUCCESS in 21m 31s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/da1cd58cdfba42a69ccd6409fd47cfa6 : SUCCESS in 25m 49s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/c175e4154b4947ada4972d74bfc13907 : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0f72d535e00c1fe8362baa523dbf227f2803b509","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 14:50:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"968c0bfffbbb6aeeaa2b84723184cf1f55d6634b","revisions":{"968c0bfffbbb6aeeaa2b84723184cf1f55d6634b":{"kind":"REWORK","_number":1,"created":"2024-10-21 12:13:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/932899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/99/932899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/99/932899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/99/932899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/99/932899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/99/932899/1"}}},"commit":{"parents":[{"commit":"96a4ea24b7f6e8ad3ebb39509648ed3397d5f15a","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/96a4ea24b7f6e8ad3ebb39509648ed3397d5f15a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-21 12:12:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-21 12:13:10.000000000","tz":540},"subject":"Skip tempest jobs for non-functional change","message":"Skip tempest jobs for non-functional change\n\nChange-Id: I58cf063dc21b1a23313c00862c28551dec4cc8eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/968c0bfffbbb6aeeaa2b84723184cf1f55d6634b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/968c0bfffbbb6aeeaa2b84723184cf1f55d6634b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
