)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~932962","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~Iba801148410e8e8a5acea55cc9cf02dc74ce80f2","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"migrate-to-noble","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-12-02 10:22:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iba801148410e8e8a5acea55cc9cf02dc74ce80f2","subject":"Update gate jobs as per the 2025.1 cycle testing runtime","status":"MERGED","created":"2024-10-22 03:00:41.000000000","updated":"2024-12-02 10:22:13.000000000","submitted":"2024-12-02 10:22:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932962-migrate-to-noble","meta_rev_id":"43a5a8a1aa2ea9d6f16df03efbb5d78a1fa99ee8","_number":932962,"virtual_id_number":932962,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-02 10:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-12-02 09:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-12-02 09:21:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-22 03:37:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-02 06:56:14.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-12-02 09:21:29.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"98f8297cf88f14ad09f41ecfaf79932ed146b30f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-22 03:00:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea2a9d089fae43f3c0f7a1b97fca6658dc326d30","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-22 03:17:07.000000000","message":"Topic set to migrate-to-noble","accounts_in_message":[],"_revision_number":1},{"id":"1899465608e9a122be7f7220ac6ae5a315a9e11c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 03:37:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8910231eb53942cbbe4a1e98a3b08618\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe4a47cdc8d9441a83c7817d64778b7f : FAILURE in 3m 49s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/51179150f65d406594a8917c0fe9d43a : FAILURE in 16m 56s\n- zaqar-tempest-plugin-mongodb-jammy https://zuul.opendev.org/t/openstack/build/601e773279cd4812b84058f0853c0114 : SUCCESS in 20m 24s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/ded412ffeada49a9b0dd39b7d4a71cc0 : FAILURE in 18m 37s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/76205fce2f4d48d78abd5a4d9d7b898e : FAILURE in 22m 53s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/263e42fa60b348f7b953d11d661397ee : FAILURE in 29m 23s","accounts_in_message":[],"_revision_number":1},{"id":"bec16375d72f5a4f88d9139bb93797f2942ed35c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-12 04:13:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"360abe6960b0fabf4b52315e352b20894b662733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 04:52:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40f4fcafdcdc40a48d8015543db9bce9\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f9cc345585074811842b1bd055f23eb5 : SUCCESS in 1m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c1c42fa12bf4f87b2abc8206d32a45f : SUCCESS in 2m 14s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/f916cd337e5f4a48b90087b436719b15 : FAILURE in 15m 20s\n- zaqar-tempest-plugin-mongodb-jammy https://zuul.opendev.org/t/openstack/build/fae9bce8b05842b0bcec5bef1e7a0e88 : SUCCESS in 21m 08s\n- zaqar-tempest-plugin-mongodb-2023-2 https://zuul.opendev.org/t/openstack/build/0442e2c8c5204eb78b88b0d5268fd933 : SUCCESS in 23m 00s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/e6dce33e3101428dad0c1a50bf8bb522 : SUCCESS in 22m 33s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/1ee79e21c6624fcfbc82a47c0f6bef23 : SUCCESS in 21m 56s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/d34f285c2d5d4750819cf4acb89c8ce1 : FAILURE in 17m 06s\n- zaqar-tempest-plugin-redis-2023-2 https://zuul.opendev.org/t/openstack/build/5964de57383844418c9669591ed0ee5e : SUCCESS in 23m 08s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/74cde62d30394fb696569651f744e0f1 : SUCCESS in 23m 14s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/90902ad109b043e8b823aecd6e569462 : SUCCESS in 19m 53s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/1f8df9a12fe84fe6b61d8f227241aad1 : FAILURE in 17m 42s\n- zaqar-tempest-plugin-swift-2023-2 https://zuul.opendev.org/t/openstack/build/154351a750ca4b46b45946232612ebfd : SUCCESS in 23m 16s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/29d48d97f90548bd8b1274866743f30e : SUCCESS in 25m 26s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/0c851bdc61bf41149990b9e1fc136989 : SUCCESS in 22m 02s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/fdd4f144e9074c5698c37234d805c6b3 : FAILURE in 33m 16s","accounts_in_message":[],"_revision_number":2},{"id":"7e7d28b4c1e13c1df7c9c087e62554db7f85ea99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-22 20:22:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d73b574283c08185c40b795e35cda59cfcb64eed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 21:10:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3aefef8b54d64104a97ff63487aeac05\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8bfa08779ea34a47973431fc782ff51b : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb135bc41ea041e6ba196ca8b8e59d66 : SUCCESS in 2m 58s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/d20546829b40408caf1475c55fc770d2 : FAILURE in 25m 47s\n- zaqar-tempest-plugin-mongodb-jammy https://zuul.opendev.org/t/openstack/build/112609e9fc8444d9bffd39cf9159b612 : SUCCESS in 21m 44s\n- zaqar-tempest-plugin-mongodb-2023-2 https://zuul.opendev.org/t/openstack/build/e03e8e70a5754c6e80306d53b68e5323 : SUCCESS in 21m 48s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/0936da1e9fe947e296f0962af1b681b1 : SUCCESS in 22m 18s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/74b5029cfa05486089f2b7fd5d20fed8 : SUCCESS in 39m 05s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/5d8718e226a8435b8fc1fbe26393e04b : FAILURE in 30m 16s\n- zaqar-tempest-plugin-redis-2023-2 https://zuul.opendev.org/t/openstack/build/05b153f766e745d6a4ad526b42ffcba2 : SUCCESS in 38m 44s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/84671969a55949cf9c13adcf5dcf3099 : SUCCESS in 40m 25s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/ff53d3c4d71d4866a554ce6b943ec784 : SUCCESS in 38m 47s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/56444e4bc0344b17a0dbfb742bdc5f56 : FAILURE in 28m 28s\n- zaqar-tempest-plugin-swift-2023-2 https://zuul.opendev.org/t/openstack/build/c136c9f443ef4ac0bb984b73d9b38d98 : SUCCESS in 37m 28s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/d0f65a40dc724e2298b0ff559b154ab9 : SUCCESS in 35m 55s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/272a58b9fff34cb4aae2831b7527e8fb : SUCCESS in 36m 39s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/60f4fd379ad6464e8fb01ba164898759 : FAILURE in 26m 58s","accounts_in_message":[],"_revision_number":3},{"id":"91256995fc0c16ad30614c597d5c1f6cbc2602c0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-02 06:56:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8c37de66461a06a3edeac5ee038cee502a69a274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 07:41:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c61bcea41dc487b931c92bb6c3b320e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c17c2c4bf85a47a2939d7a76319f4dbb : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d04b1fcd5e401fa7897d63c4b5845d : SUCCESS in 2m 26s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/093fd89041604e21b21e48deae28afe1 : SUCCESS in 22m 34s\n- zaqar-tempest-plugin-mongodb-jammy https://zuul.opendev.org/t/openstack/build/6ced42e7a79d49b2a51dc0a63198aa37 : SUCCESS in 18m 26s\n- zaqar-tempest-plugin-mongodb-2023-2 https://zuul.opendev.org/t/openstack/build/79a9ceb0cd3f4d4a8f42d54943d8b149 : SUCCESS in 35m 49s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/984749329d134f358ae820ce1846539a : SUCCESS in 21m 02s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/032c33cf5f94444bba91324d872761d2 : SUCCESS in 18m 53s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/7435b84cb8f245d5888261e2121289ad : SUCCESS in 37m 08s\n- zaqar-tempest-plugin-redis-2023-2 https://zuul.opendev.org/t/openstack/build/95180668a8f4434a9b456e58d6a7f773 : SUCCESS in 27m 20s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/80bf9b919bff44869c1c0e4e9a85a462 : SUCCESS in 20m 02s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/e7be48d729e04d9980df674801c5f690 : SUCCESS in 33m 09s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/ba28602707cb44e3a1b3eea0c6a92c96 : SUCCESS in 24m 02s\n- zaqar-tempest-plugin-swift-2023-2 https://zuul.opendev.org/t/openstack/build/045f08e0c1824affaff0de4de151ea5f : SUCCESS in 26m 19s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/6ab06e9bc5f3471a80f76c02142ee102 : SUCCESS in 38m 44s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/e2ca80a9639c4b1abfe9874c07a4a49e : SUCCESS in 18m 50s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/1a9bac9d994348479f553e3b1744fac2 : SUCCESS in 26m 48s","accounts_in_message":[],"_revision_number":3},{"id":"a1a22500c46ba46c273aec43c3356ac917282f20","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-02 09:21:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b44a972545c8b960a588962c1b39879b91f89da6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 09:22:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bd9632abd0b9b02919c5c682706c579a419a7a6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 10:22:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/559cac55b5e940e9aef0c310e8f7c837\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/43698a8678b2468eb521390b901369df : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c117de6df948e6b12a356eab9dc57d : SUCCESS in 3m 09s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/316b82fd9a7f4c15a57a499e02ae8197 : SUCCESS in 34m 04s\n- zaqar-tempest-plugin-mongodb-jammy https://zuul.opendev.org/t/openstack/build/cb1ba1925a174cd886cef2cfd8992acf : SUCCESS in 23m 34s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/96fca57884284a04887994b006785992 : SUCCESS in 37m 59s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/52a472d425944743a63b22f4c3d59f3a : SUCCESS in 55m 37s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/a2e46317606848868088b7cc32c354b3 : SUCCESS in 37m 29s","accounts_in_message":[],"_revision_number":3},{"id":"43a5a8a1aa2ea9d6f16df03efbb5d78a1fa99ee8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 10:22:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7f637985a3941637b99dc2c0f18a59fc6316abc9","revisions":{"590311a0329d138a3e1527364640819a25693466":{"kind":"REWORK","_number":1,"created":"2024-10-22 03:00:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/62/932962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/62/932962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/1"}}},"commit":{"parents":[{"commit":"96a4ea24b7f6e8ad3ebb39509648ed3397d5f15a","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/96a4ea24b7f6e8ad3ebb39509648ed3397d5f15a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-22 03:00:21.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-22 03:00:39.000000000","tz":-420},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Iba801148410e8e8a5acea55cc9cf02dc74ce80f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/590311a0329d138a3e1527364640819a25693466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/590311a0329d138a3e1527364640819a25693466"}]},"branch":"refs/heads/master"},"df67371ac860a961f01c17d41c8089c3823fbdb5":{"kind":"REWORK","_number":2,"created":"2024-11-12 04:13:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/62/932962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/62/932962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/2"}}},"commit":{"parents":[{"commit":"f0ab144aaf22d6b03add6e80ab81987e9913484f","subject":"Remove stable/2023.1 job from master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/f0ab144aaf22d6b03add6e80ab81987e9913484f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-12 04:12:17.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-12 04:13:05.000000000","tz":-480},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nAlso bump the hacking version to fix the pepe8 erorr:\n -AttributeError: \u0027EntryPoints\u0027 object has no attribute \u0027get\u0027\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Iba801148410e8e8a5acea55cc9cf02dc74ce80f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/df67371ac860a961f01c17d41c8089c3823fbdb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/df67371ac860a961f01c17d41c8089c3823fbdb5"}]},"branch":"refs/heads/master"},"7f637985a3941637b99dc2c0f18a59fc6316abc9":{"kind":"REWORK","_number":3,"created":"2024-11-22 20:22:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/62/932962/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/62/932962/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/62/932962/3"}}},"commit":{"parents":[{"commit":"cc57df23c1e884b33d06ab245f9aedf2ac368471","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/cc57df23c1e884b33d06ab245f9aedf2ac368471"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-22 20:21:28.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-22 20:22:36.000000000","tz":-480},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbuntu Jammy (for smooth upgrade from previous releases).\n\nThis commit adds a new job to run on Jammy which can be removed\nin future cycle when testing runtime test next version of Ubuntu\nas default.\n\nAlso bump the hacking version to fix the pepe8 erorr:\n -AttributeError: \u0027EntryPoints\u0027 object has no attribute \u0027get\u0027\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Iba801148410e8e8a5acea55cc9cf02dc74ce80f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/7f637985a3941637b99dc2c0f18a59fc6316abc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/7f637985a3941637b99dc2c0f18a59fc6316abc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
